)]}'
{"id":"openstack%2Fblazar~633204","triplet_id":"openstack%2Fblazar~master~I1b57210d623791a49ad5ac19f791c3b17b871ed3","project":"openstack/blazar","branch":"master","topic":"bp/placement-api","hashtags":[],"change_id":"I1b57210d623791a49ad5ac19f791c3b17b871ed3","subject":"Retry on inventory update conflict","status":"MERGED","created":"2019-01-25 13:22:54.000000000","updated":"2019-06-03 08:26:19.000000000","submitted":"2019-06-03 08:26:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633204-1559550379378-5b9cf8db","meta_rev_id":"82e4a648cc02cacf25b36e5d05fa71b5f155dc02","_number":633204,"virtual_id_number":633204,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-03 08:26:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":2,"date":"2019-06-03 07:39:36.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-05-29 05:11:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-05-28 12:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-28 16:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":1,"date":"2019-06-03 07:39:36.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-26 06:46:38.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-05-28 12:31:09.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-28 16:50:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-29 05:11:45.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-03 07:39:36.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2019-06-03 08:26:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19cdcc5528e6ad89c96d05a5b6564123609529c8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-01-25 13:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e143d4e31b704b1320bf3506ed85d17451bc7a36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 14:13:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/04/633204/1/check/build-openstack-api-ref/d2c8d37/html/ : SUCCESS in 5m 17s\n- requirements-check http://logs.openstack.org/04/633204/1/check/requirements-check/6605b13/ : FAILURE in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/1/check/openstack-tox-lower-constraints/b633c4b/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/1/check/openstack-tox-pep8/605c1f6/ : FAILURE in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/1/check/openstack-tox-py27/382da82/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/1/check/openstack-tox-py35/8dd9130/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/1/check/openstack-tox-py36/1be8665/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/04/633204/1/check/openstack-tox-docs/422df02/html/ : SUCCESS in 5m 51s\n- build-openstack-releasenotes http://logs.openstack.org/04/633204/1/check/build-openstack-releasenotes/b5c308c/html/ : SUCCESS in 4m 28s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/1/check/blazar-devstack-dsvm/0d09aa3/ : SUCCESS in 44m 15s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/1/check/blazar-devstack-dsvm-py35/fb80aeb/ : SUCCESS in 38m 04s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/1/check/openstack-tox-pylint/465b3aa/ : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c770926d1ad1f5223c0fc9c686cbe35f7e8c76d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-01-26 06:31:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"046775b57a6bf26d02063a01c4062a96965b93a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 07:20:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/04/633204/2/check/build-openstack-api-ref/06eebaf/html/ : SUCCESS in 5m 50s\n- requirements-check http://logs.openstack.org/04/633204/2/check/requirements-check/97ddd31/ : FAILURE in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/2/check/openstack-tox-lower-constraints/8963a3f/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/2/check/openstack-tox-pep8/30595b2/ : SUCCESS in 6m 14s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/2/check/openstack-tox-py27/464115c/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/2/check/openstack-tox-py35/3319f45/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/2/check/openstack-tox-py36/2bd40e9/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/04/633204/2/check/openstack-tox-docs/fcb32a2/html/ : SUCCESS in 5m 21s\n- build-openstack-releasenotes http://logs.openstack.org/04/633204/2/check/build-openstack-releasenotes/f3332bb/html/ : SUCCESS in 4m 31s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/2/check/blazar-devstack-dsvm/2fb9729/ : FAILURE in 41m 49s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/2/check/blazar-devstack-dsvm-py35/a266648/ : FAILURE in 45m 24s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/2/check/openstack-tox-pylint/9977996/ : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c38be6eb4e2df2bf67628443e8224ab58322a1ca","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-01-27 04:22:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9674311d01b5ae8821a3b8bdf4dbf070e75a2ca7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-27 05:07:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/04/633204/3/check/build-openstack-api-ref/4a04da6/html/ : SUCCESS in 6m 27s\n- requirements-check http://logs.openstack.org/04/633204/3/check/requirements-check/73b702c/ : FAILURE in 3m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/3/check/openstack-tox-lower-constraints/c340156/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/3/check/openstack-tox-pep8/ee21be9/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/3/check/openstack-tox-py27/794760d/ : SUCCESS in 7m 04s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/3/check/openstack-tox-py35/cb0e0a8/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/3/check/openstack-tox-py36/6b2f474/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/04/633204/3/check/openstack-tox-docs/53a03c6/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/04/633204/3/check/build-openstack-releasenotes/ae64f96/html/ : SUCCESS in 4m 16s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/3/check/blazar-devstack-dsvm/7759b67/ : FAILURE in 35m 12s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/3/check/blazar-devstack-dsvm-py35/935ba59/ : FAILURE in 38m 41s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/3/check/openstack-tox-pylint/37cc9e3/ : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea0ddfd257c1a47e397427c47f30233a98890691","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-01-27 14:10:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f2800d34532445981062acb3e2507262ced3294","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-27 14:56:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/04/633204/4/check/build-openstack-api-ref/acaf6cf/html/ : SUCCESS in 4m 40s\n- requirements-check http://logs.openstack.org/04/633204/4/check/requirements-check/46b3c6a/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/4/check/openstack-tox-lower-constraints/fb70b79/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/4/check/openstack-tox-pep8/ae0627c/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/4/check/openstack-tox-py27/ec83186/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/4/check/openstack-tox-py35/0fea12e/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/4/check/openstack-tox-py36/6d97698/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/04/633204/4/check/openstack-tox-docs/f3e71cd/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/04/633204/4/check/build-openstack-releasenotes/a65c9db/html/ : SUCCESS in 4m 31s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/4/check/blazar-devstack-dsvm/9193a2f/ : SUCCESS in 31m 35s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/4/check/blazar-devstack-dsvm-py35/d8af5f9/ : SUCCESS in 40m 29s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/4/check/openstack-tox-pylint/adf8632/ : SUCCESS in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4b3adf9c37d75c817294a243f8c61ccd8541b8f","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-02-26 07:22:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2c767b2589caa76c5af416c50dc6cff1c9ce8601","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-03 05:19:29.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"21777fcb2b8cc7166da2c1bb3d0c901d4d767112","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-03 05:20:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f66efde49a0fe09bec793261bca3834ea80200e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-03 06:00:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/5/check/openstack-tox-lower-constraints/2470f02/ : FAILURE in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/5/check/openstack-tox-pep8/2464f7f/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/5/check/openstack-tox-py27/fa14b01/ : FAILURE in 7m 11s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/5/check/openstack-tox-py35/776bc27/ : FAILURE in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/5/check/openstack-tox-py36/908fe97/ : FAILURE in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/5/check/openstack-tox-py37/bd1e8bc/ : FAILURE in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/04/633204/5/check/openstack-tox-docs/a33f8cf/html/ : SUCCESS in 6m 20s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/5/check/blazar-devstack-dsvm/3d30bac/ : FAILURE in 34m 32s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/5/check/blazar-devstack-dsvm-py35/49de0ce/ : FAILURE in 31m 50s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/5/check/openstack-tox-pylint/2de4bee/ : SUCCESS in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"901916ff6237ea7be417ada352fc9927ac6382ec","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-06 05:23:09.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"ea12abd03d56beea90f9790741d7a4a021e48500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 06:02:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/6/check/openstack-tox-lower-constraints/d2ce137/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/6/check/openstack-tox-pep8/6504beb/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/6/check/openstack-tox-py27/fe5e591/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/6/check/openstack-tox-py35/2d18903/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/6/check/openstack-tox-py36/5be7b72/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/6/check/openstack-tox-py37/4143619/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/04/633204/6/check/openstack-tox-docs/41a1b80/html/ : SUCCESS in 4m 52s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/6/check/blazar-devstack-dsvm/839e162/ : SUCCESS in 37m 32s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/6/check/blazar-devstack-dsvm-py35/bc51a24/ : SUCCESS in 36m 12s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/6/check/openstack-tox-pylint/a1a6404/ : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2c7bd05f35176b47af47d12015493d4837db7353","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-17 08:11:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f3aef395b2d1faebfa98f6f86fd4b837ac97f5fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-17 09:03:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/7/check/openstack-tox-lower-constraints/4327b80/ : SUCCESS in 6m 22s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/7/check/openstack-tox-pep8/fc825a4/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/7/check/openstack-tox-py27/2dc39b9/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/7/check/openstack-tox-py35/f5237cd/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/7/check/openstack-tox-py36/3f34030/ : SUCCESS in 4m 57s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/7/check/openstack-tox-py37/c94f10a/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/04/633204/7/check/openstack-tox-docs/604f65b/html/ : SUCCESS in 5m 55s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/7/check/blazar-devstack-dsvm/08983ea/ : SUCCESS in 41m 25s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/7/check/blazar-devstack-dsvm-py35/9d6cd4a/ : FAILURE in 47m 15s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/7/check/openstack-tox-pylint/4e0288a/ : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9c81da803f210cc46649d47f12584985c3d764ac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-03-18 22:46:31.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"8604061386dda526a5e42b216a10982a74d8eed9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 23:31:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/8/check/openstack-tox-lower-constraints/3d22b12/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/8/check/openstack-tox-pep8/0281606/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/8/check/openstack-tox-py27/2395661/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/8/check/openstack-tox-py35/1640664/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/8/check/openstack-tox-py36/accf853/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/8/check/openstack-tox-py37/2002a90/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/04/633204/8/check/openstack-tox-docs/097fdb6/html/ : SUCCESS in 4m 03s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/8/check/blazar-devstack-dsvm/bf458fd/ : SUCCESS in 42m 39s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/8/check/blazar-devstack-dsvm-py35/0950280/ : SUCCESS in 39m 11s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/8/check/openstack-tox-pylint/0aa897e/ : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6f7eaa25e376aae6bf3eb467aacc175046d5bb34","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-03-20 23:14:03.000000000","message":"Patch Set 8:\n\n@Tetsuro: is this critical enough to go in an RC2?","accounts_in_message":[],"_revision_number":8},{"id":"b81250ca010e39a1cf6facd5195eac5b1b3c66d9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-26 01:08:43.000000000","message":"Patch Set 8:\n\n@pierre\nI don\u0027t think so, this patch just adds an automatic retry for cases where a conflict happens. Without this automatic retry, users should retry it manually, but it will succeed without any inconsistency, which is okay IMO.","accounts_in_message":[],"_revision_number":8},{"id":"7e33214617bfd6597da81b804c85e827e4a0e28c","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-26 01:10:24.000000000","message":"Patch Set 8:\n\n..though I\u0027m good to backport this if this is merged.","accounts_in_message":[],"_revision_number":8},{"id":"757187ccaadc1eac1687302f4d47092deefcf096","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-26 01:29:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8fd4da6c3822754fc1c9d8515a902c83d84de224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 02:22:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/9/check/openstack-tox-lower-constraints/ce9eb2e/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/9/check/openstack-tox-pep8/2d33a71/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/9/check/openstack-tox-py27/cac3b30/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/9/check/openstack-tox-py35/8bea63c/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/9/check/openstack-tox-py36/14942e5/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/9/check/openstack-tox-py37/4f56d3a/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/04/633204/9/check/openstack-tox-docs/f9cc4ef/html/ : SUCCESS in 5m 06s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/9/check/blazar-devstack-dsvm/ef84a8e/ : SUCCESS in 38m 32s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/9/check/blazar-devstack-dsvm-py35/f8b3795/ : SUCCESS in 43m 01s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/9/check/openstack-tox-pylint/6a81145/ : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5efc65496d3d43899cbb6e534410856261935e6d","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-03-26 06:46:38.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nThanks! I\u0027m okay to backport this patch if it\u0027s merged in RC1-2.","accounts_in_message":[],"_revision_number":9},{"id":"9c4fd71a0abd1e90bf4c78cd9334bbe8fb94be2c","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-03-28 04:54:28.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"eb5b5c38265c87b83cf9789b48e00c3ad954afd2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-04 16:33:45.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nSuggestion within to make this a bit more robust","accounts_in_message":[],"_revision_number":9},{"id":"9a0b8b9c80bc6fed4b807d1c2b4806df185f52e8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-25 10:40:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b557790f92acd95c87fee743c000dc7cf553ba6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 11:33:45.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/10/check/openstack-tox-lower-constraints/0217d45/ : SUCCESS in 6m 47s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/10/check/openstack-tox-pep8/198c74c/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/10/check/openstack-tox-py27/6a59182/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/10/check/openstack-tox-py35/da08849/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/10/check/openstack-tox-py36/2dd30ed/ : SUCCESS in 4m 39s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/10/check/openstack-tox-py37/edebbd1/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/04/633204/10/check/openstack-tox-docs/84e5411/html/ : SUCCESS in 5m 13s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/10/check/blazar-devstack-dsvm/07e3b60/ : SUCCESS in 36m 57s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/10/check/blazar-devstack-dsvm-py35/f3a8060/ : SUCCESS in 39m 00s\n- openstack-tox-pylint http://logs.openstack.org/04/633204/10/check/openstack-tox-pylint/50e6d7a/ : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bbb88454632173b1f90c08aad1babe997f8b4b02","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-28 12:31:09.000000000","message":"Patch Set 10: Code-Review+1\n\nyeah, that seems cleaner","accounts_in_message":[],"_revision_number":10},{"id":"9a7d77337cb395790e5d75c0d2be6401ab8a1b52","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-28 16:50:10.000000000","message":"Patch Set 10: Code-Review+2\n\nThanks Tetsuro.","accounts_in_message":[],"_revision_number":10},{"id":"7e2f8cacbb26c5fc89a4f4711f694e48600262a3","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-29 05:11:45.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"e0f25bd92c92449e1b27eef1f464cff49594aa32","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2019-06-03 07:39:36.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":10},{"id":"a5a0ed2b6208951611c980535b6241f3ebd72740","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 07:39:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ba33dc04cc461b7bd56546c8e8dc9bab05011ea6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 08:26:19.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/04/633204/10/gate/openstack-tox-lower-constraints/f0e3997/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/04/633204/10/gate/openstack-tox-pep8/cc45856/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/04/633204/10/gate/openstack-tox-py27/208923c/ : SUCCESS in 7m 26s\n- openstack-tox-py35 http://logs.openstack.org/04/633204/10/gate/openstack-tox-py35/3fdf418/ : SUCCESS in 7m 23s\n- openstack-tox-py36 http://logs.openstack.org/04/633204/10/gate/openstack-tox-py36/1bb545c/ : SUCCESS in 7m 24s\n- openstack-tox-py37 http://logs.openstack.org/04/633204/10/gate/openstack-tox-py37/ac2d640/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/04/633204/10/gate/openstack-tox-docs/ca3eacd/html/ : SUCCESS in 6m 39s\n- blazar-devstack-dsvm http://logs.openstack.org/04/633204/10/gate/blazar-devstack-dsvm/f962865/ : SUCCESS in 43m 42s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/04/633204/10/gate/blazar-devstack-dsvm-py35/85e0d22/ : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":10},{"id":"7ff8b743ab45468af36a6c85636c923de25578e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 08:26:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b59d0e5a42ee6fee4728b8e8f662e5647d40b8cd","revisions":{"cc1e5f8c89a4e310d150fd1328db29ae4867d724":{"kind":"REWORK","_number":1,"created":"2019-01-25 13:22:54.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/1"}}},"commit":{"parents":[{"commit":"55945073b33d0d0d95a91ef39599250ca443fc4b","subject":"Call placement after instance delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/55945073b33d0d0d95a91ef39599250ca443fc4b"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 13:22:14.000000000","tz":0},"subject":"WIP: Retry on inventory conflict","message":"WIP: Retry on inventory conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent wich the registered. For Blazar\u0027s use case, The\ngeneration inconsistency can occur in cases another thread\nupdates the same reservation provider for a different reservation.\nThis patch changes update_inventory() to retry for such cases.\n\nThis patch also changes delete_inventory() to retry as well just in\ncase the deletion of consumer\u0027s allocation is not finished before\ncalling placement.\n\nWIP for unit tests.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cc1e5f8c89a4e310d150fd1328db29ae4867d724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cc1e5f8c89a4e310d150fd1328db29ae4867d724"}]},"branch":"refs/heads/master"},"4c9ba2b993400046d618980703b9b874086874f1":{"kind":"REWORK","_number":2,"created":"2019-01-26 06:31:09.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/2"}}},"commit":{"parents":[{"commit":"55945073b33d0d0d95a91ef39599250ca443fc4b","subject":"Call placement after instance delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/55945073b33d0d0d95a91ef39599250ca443fc4b"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-26 06:30:53.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent wich the registered. For Blazar\u0027s use case, The\ngeneration inconsistency can occur in cases another thread\nupdates the same reservation provider for a different reservation.\nThis patch changes update_inventory() to retry for such cases.\n\nThis patch also changes delete_inventory() to retry as well just in\ncase the deletion of consumer\u0027s allocation is not finished before\ncalling placement.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4c9ba2b993400046d618980703b9b874086874f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4c9ba2b993400046d618980703b9b874086874f1"}]},"branch":"refs/heads/master"},"e2b1f6c4459767dccaeff1d74b6e31b7a32b8e4c":{"kind":"REWORK","_number":3,"created":"2019-01-27 04:22:56.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/3"}}},"commit":{"parents":[{"commit":"c972ee0b93a9de75b8ac09af1c3e6fad82a0e5b7","subject":"Call placement after instance delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c972ee0b93a9de75b8ac09af1c3e6fad82a0e5b7"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-27 04:21:48.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nNow delete_inventory() in blazar\u0027s placement client is used to delete\nreservation class when the reservation ends. It is now called after\ninstance deletion, but it still raised 409 conflict error if the\ndeletion is not completed. To fix this, this patch changes it to\ndetect the 409 case and retry at some interval.\n\nThis patch also changes update_inventory() to retry for 409 conflict,\nwhich occurs in cases another thread updates the same reservation\nprovider for a different reservation.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\nCloses-Bug: #1813252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e2b1f6c4459767dccaeff1d74b6e31b7a32b8e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e2b1f6c4459767dccaeff1d74b6e31b7a32b8e4c"}]},"branch":"refs/heads/master"},"a6e2affa02d400a521f3c07b867a6a72e55e0a53":{"kind":"REWORK","_number":4,"created":"2019-01-27 14:10:06.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/4"}}},"commit":{"parents":[{"commit":"98a203b3e763bf0a55dcaf545964f62a0ff59085","subject":"Merge \"Add a release note for the List and Get Allocation APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/98a203b3e763bf0a55dcaf545964f62a0ff59085"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-27 14:09:34.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a6e2affa02d400a521f3c07b867a6a72e55e0a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a6e2affa02d400a521f3c07b867a6a72e55e0a53"}]},"branch":"refs/heads/master"},"cf08622c5090608d87c359239a3923426eb0e54f":{"kind":"REWORK","_number":5,"created":"2019-03-03 05:20:37.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/5"}}},"commit":{"parents":[{"commit":"0e96963376ad3829461fdb223b426bc452cf2d09","subject":"Merge \"Update json module to jsonutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0e96963376ad3829461fdb223b426bc452cf2d09"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-03 05:14:57.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cf08622c5090608d87c359239a3923426eb0e54f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cf08622c5090608d87c359239a3923426eb0e54f"}]},"branch":"refs/heads/master"},"2839c2fa687218fcb29129c24806f723820d63f5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-06 05:23:09.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/6"}}},"commit":{"parents":[{"commit":"9160ada28aa447875af51686568132f4e30216cc","subject":"Update placement client unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9160ada28aa447875af51686568132f4e30216cc"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-01 22:44:26.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2839c2fa687218fcb29129c24806f723820d63f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2839c2fa687218fcb29129c24806f723820d63f5"}]},"branch":"refs/heads/master"},"a505ef8537abd7589052dffd36675172de607084":{"kind":"REWORK","_number":7,"created":"2019-03-17 08:11:48.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/7"}}},"commit":{"parents":[{"commit":"81f3ba9bb399eb9f7aabe624e2cf4aeef3b3e1c6","subject":"Support adding to existing reservation inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/81f3ba9bb399eb9f7aabe624e2cf4aeef3b3e1c6"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-17 08:09:26.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a505ef8537abd7589052dffd36675172de607084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a505ef8537abd7589052dffd36675172de607084"}]},"branch":"refs/heads/master"},"b3c591db5d696ed006d81b1a5dc5deafd69eb773":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-03-18 22:46:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/633204/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/8"}}},"commit":{"parents":[{"commit":"8192ac8f06d31a702af906a05dae7bfeee74ffb2","subject":"Merge \"Expose instance reservation affinity support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8192ac8f06d31a702af906a05dae7bfeee74ffb2"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-03-18 22:46:31.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b3c591db5d696ed006d81b1a5dc5deafd69eb773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b3c591db5d696ed006d81b1a5dc5deafd69eb773"}]},"branch":"refs/heads/master"},"241002430c9254e18055242c08189054c5431097":{"kind":"REWORK","_number":9,"created":"2019-03-26 01:29:33.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/9"}}},"commit":{"parents":[{"commit":"27c3aea4fe649bc0dd7f0caaf0384c35096619d3","subject":"Add a release note for the floating IP plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/27c3aea4fe649bc0dd7f0caaf0384c35096619d3"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-16 12:24:56.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/241002430c9254e18055242c08189054c5431097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/241002430c9254e18055242c08189054c5431097"}]},"branch":"refs/heads/master"},"b59d0e5a42ee6fee4728b8e8f662e5647d40b8cd":{"kind":"REWORK","_number":10,"created":"2019-05-25 10:40:00.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/04/633204/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/633204/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/633204/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/633204/10"}}},"commit":{"parents":[{"commit":"7c2ce767e221bc9c58bdb18fe57fe6469c0077f7","subject":"Merge \"Handle SIGHUP signal for blazar-manager correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7c2ce767e221bc9c58bdb18fe57fe6469c0077f7"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-01-25 12:23:02.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-25 10:39:16.000000000","tz":0},"subject":"Retry on inventory update conflict","message":"Retry on inventory update conflict\n\nPlacement raises 409 conflict if the provided generation is\ninconsistent with the registered in placement DB. For Blazar\u0027s use\ncase, The generation inconsistency can occur in cases where another\nthread concurrently updates the same reservation provider for a\ndifferent reservation.\n\nThis patch changes update_inventory() to retry for such cases.\n\nChange-Id: I1b57210d623791a49ad5ac19f791c3b17b871ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b59d0e5a42ee6fee4728b8e8f662e5647d40b8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b59d0e5a42ee6fee4728b8e8f662e5647d40b8cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
