)]}'
{"id":"openstack%2Fblazar~668642","triplet_id":"openstack%2Fblazar~master~If951f82d6610c6ee54d594d9a8985bc78e135d2b","project":"openstack/blazar","branch":"master","topic":"bindep","hashtags":[],"change_id":"If951f82d6610c6ee54d594d9a8985bc78e135d2b","subject":"Add bindep.txt to fix unit tests","status":"MERGED","created":"2019-07-02 09:33:04.000000000","updated":"2019-07-04 14:55:39.000000000","submitted":"2019-07-04 14:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668642-1562252139236-a942bd96","meta_rev_id":"43c9babe86e30a84f2091aa52c07075474aea3d1","_number":668642,"virtual_id_number":668642,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-07-04 14:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":1,"date":"2019-07-04 11:32:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-07-04 11:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-04 14:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2019-07-02 10:30:07.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"date":"2019-07-02 10:30:12.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 14:02:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"date":"2019-07-02 10:30:02.000000000","_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 10:30:02.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-07-02 10:30:07.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2019-07-02 10:30:12.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-07-04 11:05:43.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-04 11:32:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-07-04 14:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2662862ae16564dc44de58d74ec3c9dc53c7b64e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 09:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a87a468a34e65a257d77d9e208396098a9ac040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 10:23:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/42/668642/1/check/build-openstack-api-ref/e2c3039/html/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/668642/1/check/openstack-tox-lower-constraints/21708c4/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/42/668642/1/check/openstack-tox-pep8/a854229/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/42/668642/1/check/openstack-tox-py27/88d9909/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/42/668642/1/check/openstack-tox-py35/b153e95/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/42/668642/1/check/openstack-tox-py36/d6bf3cc/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/42/668642/1/check/openstack-tox-py37/29a66b8/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/42/668642/1/check/openstack-tox-docs/0295e65/html/ : SUCCESS in 3m 21s\n- build-openstack-releasenotes http://logs.openstack.org/42/668642/1/check/build-openstack-releasenotes/696e79e/html/ : SUCCESS in 3m 55s\n- blazar-devstack-dsvm http://logs.openstack.org/42/668642/1/check/blazar-devstack-dsvm/05b7bbf/ : SUCCESS in 43m 38s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/42/668642/1/check/blazar-devstack-dsvm-py35/bb43e8d/ : SUCCESS in 39m 53s\n- openstack-tox-pylint http://logs.openstack.org/42/668642/1/check/openstack-tox-pylint/dd8e6af/ : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"511764c3068e63be1a776614291d8b92a6bf3000","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-04 11:05:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fb08cf0fe07a5dd2956d1a7d5d4433abe1b7a8d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-04 11:32:50.000000000","message":"Patch Set 1: Code-Review+1\n\nPatch looks good and tests run successful with it.\n\n@Dharmendra: e.g. here is one such failure: http://logs.openstack.org/79/669079/1/check/openstack-tox-pep8/ca768a3/job-output.txt.gz#_2019-07-04_07_53_48_794025","accounts_in_message":[],"_revision_number":1},{"id":"5933122842bd53c4322bc909fb16cd045800e825","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-04 14:02:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf approving to unbreak the gate. If anyone wants to see changes please comment, we can produce a follow-up patch.","accounts_in_message":[],"_revision_number":1},{"id":"367c58684f57469a5efc49c2eb1e30700848422c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:02:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa7dda400a61482aaa9e0f89c2836d90a0085c19","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-04 14:03:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f16dc8c415067e9379b09e8eb8380286f14a59a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:55:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/42/668642/1/gate/build-openstack-api-ref/2cc39d4/html/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/668642/1/gate/openstack-tox-lower-constraints/8954cb8/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/42/668642/1/gate/openstack-tox-pep8/924307a/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/42/668642/1/gate/openstack-tox-py27/991bce2/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/42/668642/1/gate/openstack-tox-py35/dc20d6a/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/42/668642/1/gate/openstack-tox-py36/4b0dc33/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/42/668642/1/gate/openstack-tox-py37/18a6173/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/42/668642/1/gate/openstack-tox-docs/6948391/html/ : SUCCESS in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/42/668642/1/gate/build-openstack-releasenotes/d464c4e/html/ : SUCCESS in 4m 41s\n- blazar-devstack-dsvm http://logs.openstack.org/42/668642/1/gate/blazar-devstack-dsvm/8dc3e08/ : SUCCESS in 44m 48s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/42/668642/1/gate/blazar-devstack-dsvm-py35/1464493/ : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":1},{"id":"be0a0ce19d7a055d4f159df8dded205d7c9008c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:55:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64a1d26aaf0368af4ea4773b41520a0802786fb3","revisions":{"64a1d26aaf0368af4ea4773b41520a0802786fb3":{"kind":"REWORK","_number":1,"created":"2019-07-02 09:33:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/668642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/42/668642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/42/668642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/42/668642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/42/668642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/42/668642/1"}}},"commit":{"parents":[{"commit":"5d5beba1b45ceea2f4a56097d32aba3899bb081b","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5d5beba1b45ceea2f4a56097d32aba3899bb081b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-02 09:32:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-02 09:32:10.000000000","tz":60},"subject":"Add bindep.txt to fix unit tests","message":"Add bindep.txt to fix unit tests\n\nThe unit tests fail since no mysqladmin command is installed. Add\nbindep.txt to install the needed binaries in our CI environment.\n\nAdd also the usual bindep tox environment for developers.\n\nChange-Id: If951f82d6610c6ee54d594d9a8985bc78e135d2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
