)]}'
{"id":"openstack%2Fblazar~668749","triplet_id":"openstack%2Fblazar~master~I2db61f4010593c65f92203db9cb50695939b1af1","project":"openstack/blazar","branch":"master","topic":"bp/simple-network-reservation","hashtags":[],"change_id":"I2db61f4010593c65f92203db9cb50695939b1af1","subject":"Implement network segment reservation","status":"ABANDONED","created":"2019-07-02 19:44:47.000000000","updated":"2021-09-23 16:24:31.000000000","total_comment_count":49,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b766f83b9a5221fac5e37609178815ea32bdf2f1","_number":668749,"virtual_id_number":668749,"owner":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-13 21:46:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-11-13 21:46:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 19:44:59.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-04-02 00:38:43.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-04-02 16:45:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-11-10 00:22:36.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2020-11-10 00:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05aba91e7aa0e95d76865269bf399a995edd4b7c","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-02 19:44:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a280ea6825e8e71f430dc22cce4af1a50630f39f","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-07-02 19:44:59.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"4bf9efa70d1898c38dcdd472b69e21f2a2b85970","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:38:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/1/check/test-release-openstack/640830a/ : SUCCESS in 2m 07s\n- requirements-check http://logs.openstack.org/49/668749/1/check/requirements-check/33a9deb/ : RETRY_LIMIT in 1m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/1/check/openstack-tox-lower-constraints/2b90d12/ : RETRY_LIMIT in 2m 12s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/1/check/openstack-tox-pep8/2ef7f80/ : RETRY_LIMIT in 2m 10s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/1/check/openstack-tox-py27/3037a4f/ : RETRY_LIMIT in 2m 05s\n- openstack-tox-py35 http://logs.openstack.org/49/668749/1/check/openstack-tox-py35/f5c857c/ : RETRY_LIMIT in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/1/check/openstack-tox-py36/e3416e6/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/1/check/openstack-tox-py37/93213ae/ : RETRY_LIMIT in 2m 12s\n- openstack-tox-docs http://logs.openstack.org/49/668749/1/check/openstack-tox-docs/fa30f1a/ : FAILURE in 5m 24s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/1/check/blazar-devstack-dsvm/5e51286/ : FAILURE in 35m 46s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/1/check/blazar-devstack-dsvm-py35/6c05570/ : FAILURE in 37m 11s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/1/check/openstack-tox-pylint/0c02694/ : RETRY_LIMIT in 2m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe82a1a3a62f637d766317c19516b10894bf6193","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 21:16:24.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks Jacob. I think you\u0027re missing some code for the API support, is this coming via another patch? Also, rebase on top of https://review.opendev.org/#/c/668642/1 to solve the zuul failures.\n\nWe should also approve a spec for this first.","accounts_in_message":[],"_revision_number":1},{"id":"e499f4b1c3a1b4bf567b1c24eb221170b451f61b","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-02 21:31:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2e13212de0cd8c9e8b8ee4a294c2c1f34386d07c","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-02 21:32:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"55fc801084dc5306a074eae85f80e46d1d5aa9a6","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-02 21:33:47.000000000","message":"Patch Set 1:\n\n\u003e Thanks Jacob. I think you\u0027re missing some code for the API support,\n \u003e is this coming via another patch? Also, rebase on top of\n \u003e https://review.opendev.org/#/c/668642/1 to solve the zuul failures.\n \u003e \n \u003e We should also approve a spec for this first.\n\nYep, forgot it add the new files.","accounts_in_message":[],"_revision_number":1},{"id":"60efeaa25b6f137c76de8dec418bb276a0de82cd","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-02 21:34:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c31642b4e3197707f7c8cce0f9d490d24cae3494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 22:20:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/4/check/test-release-openstack/7639f11/ : RETRY_LIMIT in 1m 33s\n- requirements-check http://logs.openstack.org/49/668749/4/check/requirements-check/d2cc363/ : FAILURE in 3m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/4/check/openstack-tox-lower-constraints/3064053/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/4/check/openstack-tox-pep8/349170f/ : RETRY_LIMIT in 1m 53s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/4/check/openstack-tox-py27/f2fd32a/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/49/668749/4/check/openstack-tox-py35/437c0c9/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/4/check/openstack-tox-py36/4690e6a/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/4/check/openstack-tox-py37/f42cbcb/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/49/668749/4/check/openstack-tox-docs/fe629d5/ : RETRY_LIMIT in 2m 17s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/4/check/blazar-devstack-dsvm/c0a03a3/ : FAILURE in 27m 38s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/4/check/blazar-devstack-dsvm-py35/7c8e4dd/ : RETRY_LIMIT in 3m 55s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/4/check/openstack-tox-pylint/a50c87e/ : FAILURE in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"13129a3786dbfca7c977239f179b8ba5119dfde0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-03 06:44:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ed2e9d6216ac6feb73965a6b1ec20908b1c1cf8e","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-08 15:56:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b4621fd945152f220585a190ef97a92c4ea18a9e","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-08 15:57:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4761ebc59e6caa1172a26bc3325583fe39e47b2d","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-08 16:00:41.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"70c68197367de15d7d53235c7f40914ec5ed7998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:38:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/6/check/test-release-openstack/eb4711e/ : SUCCESS in 2m 26s\n- requirements-check http://logs.openstack.org/49/668749/6/check/requirements-check/641670b/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/6/check/openstack-tox-lower-constraints/6d704a9/ : POST_FAILURE in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/6/check/openstack-tox-pep8/a3eae88/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/6/check/openstack-tox-py27/e4b353d/ : POST_FAILURE in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/49/668749/6/check/openstack-tox-py35/c206944/ : POST_FAILURE in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/6/check/openstack-tox-py36/4050595/ : POST_FAILURE in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/6/check/openstack-tox-py37/9f0e8ce/ : POST_FAILURE in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/49/668749/6/check/openstack-tox-docs/ec5a407/html/ : SUCCESS in 3m 33s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/6/check/blazar-devstack-dsvm/b8a89f8/ : FAILURE in 33m 59s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/6/check/blazar-devstack-dsvm-py35/7383a9d/ : FAILURE in 25m 30s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/6/check/openstack-tox-pylint/9811861/ : FAILURE in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4270991ecdbaa6b31dcc43c82811b638e5b586fe","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-22 16:08:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"58e4b3f456d0258bd1f7caea25a2b87f2f1b3759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 16:56:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/7/check/test-release-openstack/a4c8d07/ : SUCCESS in 4m 20s\n- requirements-check http://logs.openstack.org/49/668749/7/check/requirements-check/dbc0de0/ : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/7/check/openstack-tox-lower-constraints/e68a873/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/7/check/openstack-tox-pep8/61998ec/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/7/check/openstack-tox-py27/972a5f0/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/7/check/openstack-tox-py36/47321b4/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/7/check/openstack-tox-py37/1e5c161/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/49/668749/7/check/openstack-tox-docs/1b5160a/html/ : SUCCESS in 4m 37s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/7/check/blazar-devstack-dsvm/ed7d4e2/ : FAILURE in 25m 32s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/7/check/blazar-devstack-dsvm-py35/e192c4a/ : FAILURE in 40m 45s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/7/check/openstack-tox-pylint/1837154/ : FAILURE in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1edc51fe9d93ea3a8b79c3a7408ae3d22f9a1dbd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 17:13:51.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"691115bcee57c729052ea370fc1e50bc01a29eda","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-24 21:59:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"165d2ddca0b7376f8f3d47204a9ecbcbb382a893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 22:29:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/8/check/test-release-openstack/bfc44eb/ : SUCCESS in 5m 02s\n- requirements-check http://logs.openstack.org/49/668749/8/check/requirements-check/8751afa/ : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/8/check/openstack-tox-lower-constraints/8193861/ : FAILURE in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/8/check/openstack-tox-pep8/c28fec4/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/8/check/openstack-tox-py27/f680ccc/ : FAILURE in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/8/check/openstack-tox-py36/8ef8ad0/ : FAILURE in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/8/check/openstack-tox-py37/a1278da/ : FAILURE in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/49/668749/8/check/openstack-tox-docs/03df0a9/html/ : SUCCESS in 4m 34s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/8/check/blazar-devstack-dsvm/02bff25/ : FAILURE in 24m 36s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/8/check/blazar-devstack-dsvm-py35/23da319/ : FAILURE in 27m 46s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/8/check/openstack-tox-pylint/a16deb7/ : FAILURE in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"312931416abedb666a0cf6e8781c4d9062aa7771","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-24 22:35:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e167415af5b018df89b4403cdc29a6d7bb9b598d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 23:06:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/668749/9/check/test-release-openstack/a53639b/ : SUCCESS in 3m 22s\n- requirements-check http://logs.openstack.org/49/668749/9/check/requirements-check/6da656a/ : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/9/check/openstack-tox-lower-constraints/2ef6857/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/9/check/openstack-tox-pep8/492f578/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/9/check/openstack-tox-py27/f6e2a77/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/9/check/openstack-tox-py36/3a613fd/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/9/check/openstack-tox-py37/0b0f657/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/49/668749/9/check/openstack-tox-docs/b6634d4/html/ : SUCCESS in 3m 50s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/9/check/blazar-devstack-dsvm/37891d5/ : FAILURE in 24m 18s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/9/check/blazar-devstack-dsvm-py35/8ac9289/ : FAILURE in 24m 00s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/9/check/openstack-tox-pylint/314023f/ : FAILURE in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7de26378f20e6eed856b432fbca1abdd5ad232ae","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-25 14:54:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"afb66bc5f41c7c5fa5e47ad6fb00b48bec8799ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:48:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/668749/10/check/test-release-openstack/b886f64/ : SUCCESS in 4m 28s\n- requirements-check http://logs.openstack.org/49/668749/10/check/requirements-check/3b34aa0/ : SUCCESS in 2m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/668749/10/check/openstack-tox-lower-constraints/2d6754c/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/49/668749/10/check/openstack-tox-pep8/c5a59f2/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/49/668749/10/check/openstack-tox-py27/b2a523d/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/49/668749/10/check/openstack-tox-py36/44e0e0a/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/49/668749/10/check/openstack-tox-py37/26a08e9/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/49/668749/10/check/openstack-tox-docs/adb5c97/html/ : SUCCESS in 4m 29s\n- blazar-devstack-dsvm http://logs.openstack.org/49/668749/10/check/blazar-devstack-dsvm/b8e43ad/ : SUCCESS in 46m 20s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/49/668749/10/check/blazar-devstack-dsvm-py35/7655e01/ : SUCCESS in 47m 32s\n- openstack-tox-pylint http://logs.openstack.org/49/668749/10/check/openstack-tox-pylint/a4ed26f/ : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ec53e196b61e9a7f18b27f09e629c45f8e4b9836","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-02-03 20:37:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"82fe613bffe8a90787a138f1beb06aaca1d7cbdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 21:29:19.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d044cd19fdc245d4b36a990609cf5fb4 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/09fc0614937647268db518cc81206a31 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ff2a3b83a524e1a804e80908c02d9dc : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46ec4ae2e81f4726ad03f299aef50824 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfdeb89dbcb448fb6aea91606fe6c79 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d97bff70168447292ed053cce48acf8 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51427f8403be47ef80e8ddcbb6823206 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0abb31c7fd0c43029e79be8d594f60a9 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb9ab08b78e4d62a370e676ca88c959 : SUCCESS in 7m 24s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/299fbf167ead4cd0b512095a1170c96b : SUCCESS in 46m 49s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/53face8a8634463cb8b6d4ed8a15a9bd : SUCCESS in 40m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c0f8d442ea14d8890fda640819d0942 : FAILURE in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e13fc0a35cb3c9e31d71fb56298ce2ad498933a6","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-02-03 21:55:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1dc0827999513ae89db5d05398f1c6d0d97f2d11","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-02-03 22:06:33.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"5271d785194440165e728290252e3f13d7b2b8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 23:01:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af0ba5a85c484ac8bf45c001861fce2c : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02ffb8edd39e4b60bf1a968e9da5cbfb : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10fb6133df454dbbb9d3747f7524099b : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669bd497ac754f31b1c4402a43c76992 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20124115d83425c8c0dea4b8947b3f9 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/607f17efc59f46ac837e1232ea5e8d46 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ee883b9a25f4c2e9930ba82e24e538c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0fb2b33c54a46d59892138d42bcdad2 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d251a91dbf74950adad8f9e379ff245 : SUCCESS in 6m 51s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/77b3dcf8ccc64291bc82782ce0200c60 : SUCCESS in 38m 46s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/92ebb19bb88649ae98bb119f33dc60a6 : SUCCESS in 47m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21ff42e6f02947169075cba727a8e6c1 : FAILURE in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a1641123c4aabc9f125d2077086646a298e01d9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-18 09:37:16.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"fe3820f89518d89b229ee4b08b952a2392de3b03","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-02-26 16:49:17.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1aedb5ecd0436f45828c8b085911700aad4cfa99","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-02-26 16:51:38.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d849b33851a33738d9a3ed7998f9128fa621ff8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:48:18.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ecd23b837e14c799c9c228f911c11f8 : SUCCESS in 6m 21s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c52eded064647ed8591235dba7802a2 : SUCCESS in 10m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/12fdae9c1bbe4baca8554fc76bd37d3a : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/339fbdb175884ea3bc7b16cb5d02c938 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0bb06246e3e4114893904df41fbb6e7 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd84255ec1e402cad75248342e2094d : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d734cb5e3bec4283a94d8b8453155493 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72f44e63b3f4430791d11cc65b9fb58e : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f38a9cf12042c887ab32f008bc69db : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4819b16f144155b9f75463ba4451d8 : SUCCESS in 10m 08s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/7f234cf01d8345e4970e094e81d7de8c : SUCCESS in 39m 17s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3df9e2da18704cd0be05eb464271ed15 : SUCCESS in 48m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13596c2db612451d843f55cc64bde5a2 : FAILURE in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"44deec3d93d0e8f8540feaba889e8065fceccfcc","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-03-02 22:29:42.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a09e2ab10d00e4ec8e845ef7649ce571fe558add","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-03-02 22:31:13.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b7be2c3f9ab1f4227e07f14d7b7f478c64bdb621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 23:28:54.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e44af5d9dd44cc3a0dce0cfa4c76b5b : SUCCESS in 6m 12s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d4991feecdec4a8a8fa8d4078bea338c : SUCCESS in 6m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e22936deb344494af921e0eeae78be9 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cdf331a258a468d8544e1cc56f7824c : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ceb4ec7789846b2865d70b18bd91750 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05d14ff20a544bba396e1fb49b5e03b : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c368dd4c3fc145a49d8eb621b772a7cc : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c41210c7f55342689dffc7c1a7d69376 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6828e8427574d109a61937220063313 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d30e2ff97b4802a9a87e7a9ec8d1ef : SUCCESS in 9m 28s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/14605e81621d4e868c2ba2eef541f0b1 : SUCCESS in 46m 31s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3f5b8066306e4f4b8455a30b68a14d61 : SUCCESS in 39m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eace13c0526a44428e35b169b6ee7262 : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d52dbebdee4195346d879bee25fae3f7a408cc5f","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-05 20:59:21.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6e558d2545984e9d53232471508c57f7560775e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 22:14:07.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be453f0a45e34f2c8e56409d8ba83e9b : SUCCESS in 4m 24s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/105aec6362564755b25f30968a9a57b4 : SUCCESS in 6m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/639c28e91397497491390db7be373eda : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6983138bc5f42e6b30f92e0cea665e0 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26dd422bf8924db8913938a10be79680 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7f1e797f2648de9991a45142375f4f : FAILURE in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f91434eb194ef88d938b6fc900c6d0 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86e8f9f0b4b643caa7e6afe6e8a8018f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60a7b455fc7a4b37b81a06450b2abf67 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8c764521b542009285144ed5004cf7 : SUCCESS in 7m 01s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/4f2a4ec45c2a44088d21c8bf477803e3 : SUCCESS in 50m 15s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/765d3960fd894cb6baea7c2e12ab557c : SUCCESS in 49m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86df7cdb175f4a8b8ba01f927ad89c79 : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f2459aa16fa44e5a23f1b214faddf428e40c2ae9","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-06 15:52:57.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"839406c17d1b9120caac854a894d6969ac99085f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:52:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8627d10a243b42d0bbc081dc8378ee8f : SUCCESS in 4m 19s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bbb871e527f44d70bfc6f92fbcaea9e7 : SUCCESS in 7m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c47bb33e106e46c1a8a57546db6d016d : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bfeefacf4b445e3a4b5f4f71f94d878 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6aa69ea8f41d476097c2f20db719a655 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e98f45bfb6416da26db902e3b57ac6 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6583682861b742438bd591bf1564d394 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/306e44ef0bd942049d1f5a6461fdd8b3 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17635c6dc5e248f3ba4e6b1ef968407c : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ebebb82c2248de8813fdda00d61f1a : SUCCESS in 10m 18s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/78ba13d168f5419093dee8a9cecf750e : SUCCESS in 44m 08s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2f6dec7ddd45339bd48e17934bdbb1 : SUCCESS in 40m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f848faa1da9c41299dccd1ff1ff0ae21 : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"89c5f3325f98f44ca219e05774eb5b559900fb3e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-10 09:18:52.000000000","message":"Patch Set 18: Review-Priority+1","accounts_in_message":[],"_revision_number":18},{"id":"e6fec44e58b9531a0afc04473510952eda5219d7","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-03-19 07:01:39.000000000","message":"Patch Set 18: Code-Review-1\n\n(4 comments)\n\nSorry for being late. Some comments within, and can we have release note[1-2] for this?\n\n[1] https://docs.openstack.org/releasenotes/blazar/\n[2] https://docs.openstack.org/reno/latest/user/usage.html\n\nCLI documents [3] should be updated, too, whose path at `blazar/doc/source/cli`.\n\n[3] https://docs.openstack.org/blazar/latest/cli/index.html","accounts_in_message":[],"_revision_number":18},{"id":"2f8be3f34f8afb135cd34ce23ed35062d48ba2a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-23 22:07:21.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"8ea5ceacfe10b704a806c554e11a3068f0aaafa3","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-24 21:48:38.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7fc83fee23fae34612f0365d897b896dffc64ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 22:50:09.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/137ce034bca94c588717987706f42e1b : SUCCESS in 3m 47s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a57734f0c40144e2b5897df81dbd2c3a : FAILURE in 4m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9f5e16eab5949b1827a6628031e3316 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/569ffccabcf04ce78d49a0bd317f51e4 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e7dd04ba4c34826868eb7fbe066136a : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbdf5195065b4827b68d6c2cf8aa005c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68aafb17fc94180b57dc2252c25ea3e : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fabf50d6a5d34669a191b8b0cfed58be : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41a52eb058d4d05a991a6f5359f40d5 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd957e82fec4f81ab6594d6f9f8fd79 : SUCCESS in 10m 30s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/a10c24ac8bb2403d8a3f24b6a745c88b : SUCCESS in 53m 37s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/fe65d1c99e774084abe4282199f861c9 : SUCCESS in 44m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/529d620e9c5c4826880a2f092c3151c3 : FAILURE in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4590346ba1953b54bc0c9bc87dee46d0174e6812","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-25 14:52:21.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"bf30a09111f00f9efe1348da09aac871d5b94fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:40:11.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eec32ee04aeb444fb77c684fe32b8152 : SUCCESS in 4m 10s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6786c1a16a744d989b702cbc6d54e065 : FAILURE in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d006d4892452495c989a0e08018da7f7 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad84f069a0f24b3eafeb1e119d16c8c9 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed7382d8197043a1829f0d04fb1376e0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6760e50472e34e88b88c11f0444d19cc : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9524baeeb632446b9680bc0d36addc45 : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/813116cf91d749df9f42516f0079e23a : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a54e0218bb6462fb62cfc5456755217 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67f63fcbd9f43d58af555e0c362750e : SUCCESS in 8m 42s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/6243dd18b0cc4160ac731ab215823ea4 : SUCCESS in 42m 40s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/887679f74d1246c9a3a644af97504407 : SUCCESS in 39m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7da8f140d1134c69828091e7463756aa : FAILURE in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ecc92739b9b5f07312cde96e4b4ab3e66c2785af","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-25 16:48:35.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6ddff50e73939556d963c6f4b65e57996dc707a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 17:37:32.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b4ec5c9128647e98f1ab2d4de0ddaab : SUCCESS in 3m 56s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/29684fca47a147b0903ecafd542cb106 : FAILURE in 7m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2666ab84a5cd454e9595a9bd0c9a7d6a : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d3c84c794b0489d9a8ac9886ff1c089 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b8ab6629c05408ab0d41864ad0bde14 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa761ca733d5428fa91ff1655518e4b9 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d34c7ce0a44e7185b12d176ffe6981 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/675a20ef32f84db09c8796a5841948e3 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25794db5dc2c4f949768f0d637ba24bf : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dc31cb3dd64614b6482b36e087c2ca : SUCCESS in 9m 12s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/000f243ad0af4fc5b0e2a15a60bf6dc5 : SUCCESS in 44m 24s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/663074686cf9404fbb99d76f83d11672 : SUCCESS in 43m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/577fb0a4f378438899e1e85d5a46080b : FAILURE in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"2142c870a767ed012b84a2e9a6935fef122bfd5d","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-25 18:02:55.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"c06b7597c566aa0112153b3dab4f8b95a07c2a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:04:32.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac4b66c3871445f197f0dba8994a67b3 : SUCCESS in 4m 30s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c605b062115b47c7846fd7cc15346d37 : SUCCESS in 5m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afb41115e0af4626a8eed7d94c7a6b8c : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac64ca4bd69431db9dc4564d2fe3eda : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d3c914235b458c8a7656abcc4062f7 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc39c685a989457dbe5eecb7124141c4 : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ee018be3064dac9488e0675555df39 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba57247d3c1740d5a7f5a5b8483d48f1 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdfb1e36ad204d13bf69563d3ca648ef : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24746c2b55cf4dc38e05c0e163c33c04 : SUCCESS in 8m 25s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/daaa62cf8f4b4bc2b3735ab752982706 : SUCCESS in 39m 01s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/d63eea980750406d80fce57fac34183c : SUCCESS in 56m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51d953dc87064251a7ad1be33fa2f048 : FAILURE in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"532f00b89a260b9bcc311186dba8ec3a27c2ef5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-26 15:40:56.000000000","message":"Patch Set 22:\n\n(16 comments)\n\nJust a quick pass for now, more thorough review to come…","accounts_in_message":[],"_revision_number":22},{"id":"6bc2bf141b83af23d5bc0767885d574d502fa469","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-26 15:47:45.000000000","message":"Patch Set 22: Code-Review-1","accounts_in_message":[],"_revision_number":22},{"id":"5fbc07c780da5df7b9311d378600822d5037938b","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-03-30 19:34:30.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"8feb2295a3b38c6cbd9d38748ea8916ac1b1a73d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 20:18:56.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d84535ae17fa41eaa7c45f00c87eec48 : SUCCESS in 4m 27s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0c38df4a986f4947b1d7053105f889e0 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1b4585e879d498190c58f99aee7e14c : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d27d2c6d35d04363833a2ac723af5632 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4af26eaf571460487f2a89741ce9ab7 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c20b3352a96442339a53edfc9773b831 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/688d094cc5cf4467a96c6d46bb0eea39 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d18630208e1d434ca6751fdcf2b878d5 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b84d15af6bb4d4798ed558c34fcc784 : SUCCESS in 11m 40s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/7dddaa17e8124ed3a09051fc1957a58b : SUCCESS in 36m 34s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/fe22586f4d204bf6b0e0ba09f5650c77 : SUCCESS in 39m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc37b7ad21ee43139695e6e75865c36a : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3647214ebd0697e2226d0cbc2c91627a2093a746","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-04-02 00:38:43.000000000","message":"Patch Set 23: Code-Review-1\n\nThank you for addressing my inline comments.\nWhy not have the release note[1-2]?\n\n[1] https://docs.openstack.org/releasenotes/blazar/\n[2] https://docs.openstack.org/reno/latest/user/usage.html\n\nCLI documents [3] could be updated, too, whose path at `blazar/doc/source/cli`.\n\n[3] https://docs.openstack.org/blazar/latest/cli/index.html","accounts_in_message":[],"_revision_number":23},{"id":"4c0de4f82b4fd218b794f22599e3e405a51ffe0d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 16:45:03.000000000","message":"Patch Set 23: Code-Review-1\n\nI tried this change and failed to create a network due to DB exceptions. There are still mismatches between segment_id and segmentation_id.\n\nIt also shows that unit tests are not sufficient for being confident in this code.","accounts_in_message":[],"_revision_number":23},{"id":"98cb7ed79804776a28d8e6023d538bd95729632c","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-04-09 16:16:14.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"b138d04bdde4fbada64722bcdf3619c7a864eee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:56:40.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9276727ed3284663913b047e49add38b : SUCCESS in 3m 33s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf2d9d4944204e10bc5c056fdfa0c13b : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb6ca0382e53488eb9e0c4bb195163cb : FAILURE in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/793fe25037fe44579d0637bbd00355e4 : FAILURE in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560f03403c674d77beb08b1522d9ea43 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91682f3711e14492bf8ded8b260d6d40 : FAILURE in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18ea01864936410ea5deadf10a42024c : FAILURE in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b10cb703c97c4535a35318ae1a114f32 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcabdb2dabfe44b08e29e8ba560e2f90 : SUCCESS in 9m 14s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/701946d509a1417eb6c0d368a9bae97f : SUCCESS in 37m 27s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/e545e0f64d554f838c2296737b252403 : SUCCESS in 35m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9cf71c2cc5c9492b808ee916823bc4f5 : FAILURE in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"dfd1243cbdc34c02bd6fc7dc89cf5bd178669a08","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-11-10 00:16:52.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"5023cd319dd37f6b97bbd9b84a920ba1130bfd82","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-11-10 00:22:36.000000000","message":"Patch Set 25: Code-Review-1\n\nJust as a FYI, I\u0027m picking this back up. I still need to get API ref docs and the CLI instructions added here.","accounts_in_message":[],"_revision_number":25},{"id":"2f7c85b0e83d27e202eb16dc3ae1c193ff566b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 00:50:53.000000000","message":"Patch Set 25: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b123c6100184446688721dddd526fadc : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/63eac63895de48ffa332e30a38a78fa7 : FAILURE in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08eaec94f4b49fc8f739d0ee780cb31 : POST_FAILURE in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff601ca0f10f44df9d4898d32370ca4b : POST_FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9bd9be5554b410f99fc3bc537438e48 : FAILURE in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3dd6e687d64d88b5d5a0fe59f1f1f4 : POST_FAILURE in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f3d7c9543d46b3bcac6ad9ff481c23 : POST_FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc50510ac90d46db8994bb35f606c17a : POST_FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffa981efaf147ab8ea1ee3961bea7f2 : FAILURE in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c7b3b2d55642419e52a759839b9959 : SUCCESS in 3m 56s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/198641c264674fa6b60c95b95ff3a668 : FAILURE in 28m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/17a7617dab1340c2802a1b2bb7afb9cb : FAILURE in 23m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac8733dc92074cd0b54d829a99ff48d1 : FAILURE in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"228d852136f272e1242bc716130f9560aab9ecc8","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2020-11-13 21:46:45.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"60b45623fdad3bfd9dd15c402151631aecec495b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 22:23:24.000000000","message":"Patch Set 26:\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6a56f9b297a4de3847f3455d42bc7fd : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/105eb98787404b249a90f5d0242ec58d : FAILURE in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57b2d4b930264526bd44e70252876371 : POST_FAILURE in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/978016ac010a46cb931ac49721f95b37 : POST_FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7f7180ddf94d2aa2618b73b36e1709 : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95e232a6955742ce890d59c616783681 : POST_FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0affce74a60e4af59e573367b3fd5a12 : POST_FAILURE in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7b8cc977be342858bab6f1ea8d6e062 : POST_FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e66363fa7f4e11a88e21d1eb4cfc2a : FAILURE in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d50ddbb7c4b347e5aab7407d0220086e : SUCCESS in 3m 28s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8a30b0bbface431e9a8874bdf01920b0 : FAILURE in 30m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/fc633e2d089f4987a44ae7b643f16c14 : FAILURE in 23m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/110e49dad8f9427daa2704d57b4bcd84 : FAILURE in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"b766f83b9a5221fac5e37609178815ea32bdf2f1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-09-23 16:24:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"d4677dd759cf42c32d424e35e4586e4305ea7a1d","revisions":{"bfb37df1902316a04516580def931e33fcbb17a3":{"kind":"REWORK","_number":1,"created":"2019-07-02 19:44:47.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/1"}}},"commit":{"parents":[{"commit":"5d5beba1b45ceea2f4a56097d32aba3899bb081b","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5d5beba1b45ceea2f4a56097d32aba3899bb081b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-02 18:56:28.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bfb37df1902316a04516580def931e33fcbb17a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bfb37df1902316a04516580def931e33fcbb17a3"}]},"branch":"refs/heads/master"},"62950035f3b26a79d6d10a027544eb05c0c22526":{"kind":"NO_CHANGE","_number":2,"created":"2019-07-02 21:31:09.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/2"}}},"commit":{"parents":[{"commit":"5d5beba1b45ceea2f4a56097d32aba3899bb081b","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5d5beba1b45ceea2f4a56097d32aba3899bb081b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-02 21:30:11.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/62950035f3b26a79d6d10a027544eb05c0c22526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/62950035f3b26a79d6d10a027544eb05c0c22526"}]},"branch":"refs/heads/master"},"da459a40d95479718e3985abe0e82aa9e4e8d627":{"kind":"REWORK","_number":3,"created":"2019-07-02 21:32:59.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/3"}}},"commit":{"parents":[{"commit":"5d5beba1b45ceea2f4a56097d32aba3899bb081b","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5d5beba1b45ceea2f4a56097d32aba3899bb081b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-02 21:31:22.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da459a40d95479718e3985abe0e82aa9e4e8d627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da459a40d95479718e3985abe0e82aa9e4e8d627"}]},"branch":"refs/heads/master"},"a689b4ef0304c42e52c70db02961943b10063d63":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-02 21:34:56.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/4"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","date":"2019-07-02 21:34:56.000000000","tz":0},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a689b4ef0304c42e52c70db02961943b10063d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a689b4ef0304c42e52c70db02961943b10063d63"}]},"branch":"refs/heads/master"},"728fb5ad7a54a69ea961d0cfc60707807dfdcef1":{"kind":"REWORK","_number":5,"created":"2019-07-08 15:56:21.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/5"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-08 15:56:08.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/728fb5ad7a54a69ea961d0cfc60707807dfdcef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/728fb5ad7a54a69ea961d0cfc60707807dfdcef1"}]},"branch":"refs/heads/master"},"d6d88760585c28e70fe551b479a2173fc1d9bcc1":{"kind":"REWORK","_number":6,"created":"2019-07-08 15:57:31.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/6"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-08 15:57:21.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d6d88760585c28e70fe551b479a2173fc1d9bcc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d6d88760585c28e70fe551b479a2173fc1d9bcc1"}]},"branch":"refs/heads/master"},"898aed288766bf5481c690d52ae5180ecaa762ff":{"kind":"REWORK","_number":7,"created":"2019-07-22 16:08:40.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/7"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-22 16:08:33.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/898aed288766bf5481c690d52ae5180ecaa762ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/898aed288766bf5481c690d52ae5180ecaa762ff"}]},"branch":"refs/heads/master"},"181b4c1a54142698439143d13a818c943b7f1f49":{"kind":"REWORK","_number":8,"created":"2019-07-24 21:59:33.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/8"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-24 21:59:06.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/181b4c1a54142698439143d13a818c943b7f1f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/181b4c1a54142698439143d13a818c943b7f1f49"}]},"branch":"refs/heads/master"},"d6b11215ccd15e80ad53c4d28682da190498499e":{"kind":"REWORK","_number":9,"created":"2019-07-24 22:35:53.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/9"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-24 22:35:42.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d6b11215ccd15e80ad53c4d28682da190498499e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d6b11215ccd15e80ad53c4d28682da190498499e"}]},"branch":"refs/heads/master"},"211334d3d349d48372e30f67667874ba2e28c261":{"kind":"REWORK","_number":10,"created":"2019-07-25 14:54:33.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/10"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 14:54:24.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/211334d3d349d48372e30f67667874ba2e28c261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/211334d3d349d48372e30f67667874ba2e28c261"}]},"branch":"refs/heads/master"},"f9479bd74c17d2f322b951a1dc0990afade524e8":{"kind":"REWORK","_number":11,"created":"2020-02-03 20:37:26.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/11"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-02-03 20:37:19.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n\nImplement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f9479bd74c17d2f322b951a1dc0990afade524e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f9479bd74c17d2f322b951a1dc0990afade524e8"}]},"branch":"refs/heads/master"},"acecfe376e64cadd6b05d91260e27ddccd342a88":{"kind":"REWORK","_number":12,"created":"2020-02-03 21:55:14.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/12"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-02-03 21:55:08.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n\nImplement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/acecfe376e64cadd6b05d91260e27ddccd342a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/acecfe376e64cadd6b05d91260e27ddccd342a88"}]},"branch":"refs/heads/master"},"cbe4209cfc68bcfb3f3f4307673471ef8b879647":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2020-02-03 22:06:33.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/13"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-02-03 22:06:12.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cbe4209cfc68bcfb3f3f4307673471ef8b879647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cbe4209cfc68bcfb3f3f4307673471ef8b879647"}]},"branch":"refs/heads/master"},"703d2fbcb8302b0752d7665f0602fb9ee80d15f6":{"kind":"REWORK","_number":14,"created":"2020-02-26 16:49:17.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/14"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-02-26 16:49:10.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/703d2fbcb8302b0752d7665f0602fb9ee80d15f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/703d2fbcb8302b0752d7665f0602fb9ee80d15f6"}]},"branch":"refs/heads/master"},"3a3d777d9190f8a89e9d6978d5c377f9cf4d4b7d":{"kind":"REWORK","_number":15,"created":"2020-02-26 16:51:38.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/15"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-02-26 16:51:31.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3a3d777d9190f8a89e9d6978d5c377f9cf4d4b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3a3d777d9190f8a89e9d6978d5c377f9cf4d4b7d"}]},"branch":"refs/heads/master"},"8c0cfca25286d1e92160621dae2e4b78922ae638":{"kind":"REWORK","_number":16,"created":"2020-03-02 22:29:42.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/49/668749/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/16"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-03-02 22:27:33.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8c0cfca25286d1e92160621dae2e4b78922ae638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8c0cfca25286d1e92160621dae2e4b78922ae638"}]},"branch":"refs/heads/master"},"2deabc18a4179792242ff794a5bb1a21d04d86de":{"kind":"REWORK","_number":17,"created":"2020-03-05 20:59:21.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/17"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-05 20:59:19.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2deabc18a4179792242ff794a5bb1a21d04d86de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2deabc18a4179792242ff794a5bb1a21d04d86de"}]},"branch":"refs/heads/master"},"525c3dddbfbc569c5086fd04e5884050ab496c79":{"kind":"REWORK","_number":18,"created":"2020-03-06 15:52:57.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/18"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-06 15:52:49.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete.\nAllocations will need to be added in future.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/525c3dddbfbc569c5086fd04e5884050ab496c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/525c3dddbfbc569c5086fd04e5884050ab496c79"}]},"branch":"refs/heads/master"},"37403cb75ccf58b09e8588bbb084dba0a47e755c":{"kind":"REWORK","_number":19,"created":"2020-03-24 21:48:38.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/19"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-24 21:48:15.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/37403cb75ccf58b09e8588bbb084dba0a47e755c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/37403cb75ccf58b09e8588bbb084dba0a47e755c"}]},"branch":"refs/heads/master"},"87d0469eb7bc637a86d8f6ed304690b1b1ba9a76":{"kind":"REWORK","_number":20,"created":"2020-03-25 14:52:21.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/20"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-25 14:52:09.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/87d0469eb7bc637a86d8f6ed304690b1b1ba9a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/87d0469eb7bc637a86d8f6ed304690b1b1ba9a76"}]},"branch":"refs/heads/master"},"6968308fe35b1120fd776791456299aad7747593":{"kind":"REWORK","_number":21,"created":"2020-03-25 16:48:35.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/21"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-25 16:48:27.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6968308fe35b1120fd776791456299aad7747593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6968308fe35b1120fd776791456299aad7747593"}]},"branch":"refs/heads/master"},"d9427732b749969727f11a0649bd4d56b6d67b8f":{"kind":"REWORK","_number":22,"created":"2020-03-25 18:02:55.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/22"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-25 18:02:47.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d9427732b749969727f11a0649bd4d56b6d67b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d9427732b749969727f11a0649bd4d56b6d67b8f"}]},"branch":"refs/heads/master"},"90926f2fd3d87d1e1cffd44e59c3c1786d385711":{"kind":"REWORK","_number":23,"created":"2020-03-30 19:34:30.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/23"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-03-30 19:34:21.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/90926f2fd3d87d1e1cffd44e59c3c1786d385711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/90926f2fd3d87d1e1cffd44e59c3c1786d385711"}]},"branch":"refs/heads/master"},"cfeca2e6ec67d8fbf50fa12f51a577d26825c2e9":{"kind":"REWORK","_number":24,"created":"2020-04-09 16:16:14.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/24"}}},"commit":{"parents":[{"commit":"71b064b36a9a6ab938b70ca32d16ed3e055ae36f","subject":"Update contributor docs for Python 2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/71b064b36a9a6ab938b70ca32d16ed3e055ae36f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-04-09 16:16:06.000000000","tz":-300},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nAdds network resource to blazar. Functionally,\nusers can reserve network segments by vlan tag.\n\nResource implements list, get, create, update, delete,\nand get/list allocations.\n\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cfeca2e6ec67d8fbf50fa12f51a577d26825c2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cfeca2e6ec67d8fbf50fa12f51a577d26825c2e9"}]},"branch":"refs/heads/master"},"28cc735d9e421eb724c54ce44415bbfc23499fe7":{"kind":"REWORK","_number":25,"created":"2020-11-10 00:16:52.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/49/668749/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/25"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-10 00:16:42.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nThis implements the network reservation spec, introducing a new plugin\n\"network\" to allow reserving network segments.\n\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/28cc735d9e421eb724c54ce44415bbfc23499fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/28cc735d9e421eb724c54ce44415bbfc23499fe7"}]},"branch":"refs/heads/master"},"d4677dd759cf42c32d424e35e4586e4305ea7a1d":{"kind":"NO_CHANGE","_number":26,"created":"2020-11-13 21:46:45.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/49/668749/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/49/668749/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/49/668749/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/49/668749/26"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-01-07 13:47:53.000000000","tz":0},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2020-11-13 21:46:17.000000000","tz":-360},"subject":"Implement network segment reservation","message":"Implement network segment reservation\n\nThis implements the network reservation spec, introducing a new plugin\n\"network\" to allow reserving network segments.\n\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nCo-Authored-By: Jason Anderson \u003cjasonanderson@uchicago.edu\u003e\nImplements: blueprint basic-network-plugin\nChange-Id: I2db61f4010593c65f92203db9cb50695939b1af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d4677dd759cf42c32d424e35e4586e4305ea7a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d4677dd759cf42c32d424e35e4586e4305ea7a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
