)]}'
{"id":"openstack%2Fblazar~672763","triplet_id":"openstack%2Fblazar~master~I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66","project":"openstack/blazar","branch":"master","topic":"floatingip-reservation-update","hashtags":[],"change_id":"I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66","subject":"Add update_reservation to floating IP plugin","status":"MERGED","created":"2019-07-25 15:56:44.000000000","updated":"2019-09-13 11:40:18.000000000","submitted":"2019-09-13 11:39:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672763-1568374752265-b9dbb748","meta_rev_id":"9b57676d8212ae11c517c40a71e22a6206539d37","_number":672763,"virtual_id_number":672763,"owner":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-09-13 11:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-13 10:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-13 10:48:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 16:08:54.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-09-13 07:07:18.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-13 10:48:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-13 11:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f5bcccada2ec2c87b548c859367c7716776ef7a","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-25 15:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b803fb4e68f77e97ad4035262f71de6469d149c0","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-07-25 15:58:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d692767aa5362240992944690e6d013464b86b7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:55:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/63/672763/2/check/openstack-tox-lower-constraints/08b45ba/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/63/672763/2/check/openstack-tox-pep8/6894c4b/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/63/672763/2/check/openstack-tox-py27/3856263/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/63/672763/2/check/openstack-tox-py36/cbc9614/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/63/672763/2/check/openstack-tox-py37/86a3e33/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/63/672763/2/check/openstack-tox-docs/37e1655/html/ : SUCCESS in 6m 22s\n- blazar-devstack-dsvm http://logs.openstack.org/63/672763/2/check/blazar-devstack-dsvm/67f377c/ : SUCCESS in 46m 41s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/63/672763/2/check/blazar-devstack-dsvm-py35/b66afc9/ : SUCCESS in 46m 24s\n- openstack-tox-pylint http://logs.openstack.org/63/672763/2/check/openstack-tox-pylint/43e955f/ : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86954550417216ee321aa4b449124f39a01f0aaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-06 14:26:35.000000000","message":"Patch Set 2: Code-Review-1\n\nThis includes Chameleon usage enforcement code.","accounts_in_message":[],"_revision_number":2},{"id":"6e6e728186952ba36952387c2f6d71427e5eb036","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-09-09 17:33:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1407a706f7c5690914312d7ddede18eb2d52db54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:30:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1e92571dcd148bc84e78ec2f7e9e695 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec80be7b8014cdb8f1cbfb7744fe435 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d452d9e3c77b41cc9a2e7897bb1f5de0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6074ab7aa6542b49bc1724277848230 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2386a2ab32894aa5a86d549ce20f95d1 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b93495277d4f57916c85bf9db4d686 : SUCCESS in 4m 54s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/ad99240002b14ce5afeeecf417e387f3 : SUCCESS in 45m 58s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/2fe68b92c9394e54b0c4a0de5cabefa7 : SUCCESS in 42m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acf1555ba1b748ad9943f21584d8b7f8 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24d6e1fd12be4d9acd47963d62e6e6d42215b6b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 17:35:46.000000000","message":"Topic set to floatingip-reservation-update","accounts_in_message":[],"_revision_number":3},{"id":"dfe7eeac346f6b670478c2a5e2adee2abbc4de12","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 17:45:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)\n\nThanks Jake for the initial patch. I\u0027ve found a number of issues with it (see comments), I will push an updated patch.","accounts_in_message":[],"_revision_number":3},{"id":"827a151e16d1a9550be81ed0e878e5b7821762e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 17:49:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b712f614539db8e3d78724c598519882320eb810","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 18:19:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d185e80128aa8ec4cc5be5a3a507d89f4a31f459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 19:11:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e65303231706453d928427ead83086b4 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31794cb35d6a472e94b53986f5b06769 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1eb77727fbf545b5b1141c1433f7043a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1b1fd2baac14c10ab3cff237f91f741 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a19af2c637b4814ace9848d3f53da0c : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b66d96edae4645b0d0ad3331c92e8e : SUCCESS in 5m 27s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/5ec92ec9f460489fa8c598477ddd51f7 : SUCCESS in 45m 13s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/008fa7618aae4d33a50a42452ba52b21 : SUCCESS in 36m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d28b1f18b19a47cdbe49caf701a1b320 : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dce1321cea05462efe62646990c5cc8e864b934c","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-09-10 20:32:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1655fd78ae24f47388607199ada82effbd4304b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-11 16:00:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3711b0cd7a72e7225b4c1c31dfe231ad7c891cf3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-11 16:01:41.000000000","message":"Patch Set 5:\n\nAdded a release note.","accounts_in_message":[],"_revision_number":5},{"id":"049d637db94ba3da13c1581e0551d354b90d27b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 17:26:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b1466b76b65491d933993fbd0fe03f0 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08d624965594f55804a31bad94e7b0f : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/089ee88450b446a3b5c6dbaf00ba9ee9 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c369b04938d4238aed555e6ec93f0fb : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09373ce4bf8341eb8a980e7f4ae8306d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518177dc3cf14590b6fd5367f68e74d8 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5f088d53f9847089a7a80c847e38240 : SUCCESS in 4m 03s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/97394d2c200f43a9a360549cf6cc91cb : SUCCESS in 43m 49s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/d6481a5660c1476ba62da02c6864785b : SUCCESS in 43m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa027cae33d74fe4866d63de8452a17a : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ffdb344c1e7e148364ca2716b4d0af2a1d8d883f","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-09-11 18:18:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d7f12fb0d77d212b02755490ef4d42ce4fc27be9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-12 11:41:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nThis is calling a function that does not exist.\n... Can we have tests to exercise this?","accounts_in_message":[],"_revision_number":5},{"id":"b4eb3c85f4fbab7e46dc4257b85766ed4d5b346f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-12 11:54:50.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8b8a9e6ee8cb40ceea7dd7ef32a9cef4918f5765","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-12 13:56:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0aa4d41b2e1d45a94e0ebef055bc9f50d4f2ee24","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-12 13:56:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0523487c5a8c654b49660a1fb8917b777c2e31f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-12 13:57:59.000000000","message":"Patch Set 6:\n\n@Tetsuro: fixed wrong method call and added some tests","accounts_in_message":[],"_revision_number":6},{"id":"32ec0b2181d430fd9e88503221b540b344a5855c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:52:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d3b50056ab14375958f201f32970ae8 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389da7a4e4a3457c8ccd7ee8c086ad90 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7772f7cc397414e8d19fd3cc2b853e2 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbd0abd50ea1421282d73f29e3e372dd : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/349d61d580254dfcbd43aa803433c577 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25890bff10be4085a3b52024c29d3bd0 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1863011b444d463b8443d2bd811828a8 : SUCCESS in 3m 45s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/eb49df129c434601b995de5a784f2561 : SUCCESS in 47m 45s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/8a906e2f1df3452394a648c22da927dd : SUCCESS in 39m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14bc01b6cc1741b0ba0bb5fd05544acd : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"38628f11b99159f9d51bc4194c86a3c2ae75a1f9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-13 07:07:18.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"02d19a10367570fad164696b44cfab60db15bc90","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-13 09:58:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a69396ecd1f1a08318a8f8e7d17b827861b79d94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:46:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0849db839ef462fb7ff159b454cd1d0 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1219e6607a64fa8b14475652672ede0 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17227597effe4b70b9236f8c87cad686 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04047170934142deb361635da9818036 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf09cf62612f4372bccd56f7619add08 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d659dc7f7434449aef5acf66af28002 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df8ee66622e84691a4a44fbfa3c1bcc4 : SUCCESS in 3m 51s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/8de1aa79993d4d3897ef5546790e5ac5 : SUCCESS in 45m 56s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/d13d016e68684754a0b053343ee26160 : SUCCESS in 37m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac15e00628ee403da0b8e359145f2ad2 : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4aaed1856882612048856e5897fce07db7827b56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-13 10:48:42.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nThanks Tetsuro, fixed the nit as well.","accounts_in_message":[],"_revision_number":7},{"id":"a7aaf097712eeafd4d4184a8a30902240605d61f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:48:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2e8e3cc1e14f4d26d1ae352d1ee9b76315e9973b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:39:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9fb2a67af86a8fd9c5a96866924cdd43e96cc4f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:39:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ebc5093bd8442c8dece50367dd3bb7 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c4f2ba13944f669dd43ec37ae8c249 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac1e65d098254e6b83814a57ae837250 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f12e7a73569b4aadb9193ee54c9d4e30 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adf590d398fe45da9251d81b68de7ba5 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef04bcba9dc1460d95f0f8f65955aaf6 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dee070513889416d8a6a7b99ab7d666f : SUCCESS in 4m 02s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/60f5251d47b0423783b2a55ac9364928 : SUCCESS in 42m 46s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/737313b83cdd4ec49e820e42068cd899 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":7},{"id":"9b57676d8212ae11c517c40a71e22a6206539d37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:40:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f162f0763642308805a8fa39d77f3d : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239fa823696d43f5ad83cd9bb98b3c62 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3137ffc84d7791aa5daa5f12dd2af68386812505","revisions":{"e4c54120211e4e909397d7fb46ebb79957111087":{"kind":"REWORK","_number":1,"created":"2019-07-25 15:56:44.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/63/672763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/1"}}},"commit":{"parents":[{"commit":"125f53bf483349898138a7cb945832bc2efaad5b","subject":"Merge \"Delete specs from blazar repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/125f53bf483349898138a7cb945832bc2efaad5b"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:51:40.000000000","tz":-300},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:55:45.000000000","tz":-300},"subject":"Add update_reservation to floating ip plugin","message":"Add update_reservation to floating ip plugin\n\nUpdate reservation functionality was not implemented\nin original plugin. Consequently, any mixed resource\nlease containing a floating ip resource will fail when\nlease is updated.\n\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e4c54120211e4e909397d7fb46ebb79957111087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e4c54120211e4e909397d7fb46ebb79957111087"}]},"branch":"refs/heads/master"},"812350a545c4b3ca664ee7c08215d0d7342eabad":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-25 15:58:35.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/63/672763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/2"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:51:40.000000000","tz":-300},"committer":{"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:58:35.000000000","tz":0},"subject":"Add update_reservation to floating ip plugin","message":"Add update_reservation to floating ip plugin\n\nUpdate reservation functionality was not implemented\nin original plugin. Consequently, any mixed resource\nlease containing a floating ip resource will fail when\nlease is updated.\n\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/812350a545c4b3ca664ee7c08215d0d7342eabad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/812350a545c4b3ca664ee7c08215d0d7342eabad"}]},"branch":"refs/heads/master"},"75a188e1d78570d8f6bcea66490cbc32a9bea0c6":{"kind":"REWORK","_number":3,"created":"2019-09-09 17:33:32.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/63/672763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/3"}}},"commit":{"parents":[{"commit":"125f53bf483349898138a7cb945832bc2efaad5b","subject":"Merge \"Delete specs from blazar repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/125f53bf483349898138a7cb945832bc2efaad5b"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:51:40.000000000","tz":-300},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-09-09 17:33:24.000000000","tz":-300},"subject":"Add update_reservation to floating ip plugin","message":"Add update_reservation to floating ip plugin\n\nUpdate reservation functionality was not implemented\nin original plugin. Consequently, any mixed resource\nlease containing a floating ip resource will fail when\nlease is updated.\n\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75a188e1d78570d8f6bcea66490cbc32a9bea0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75a188e1d78570d8f6bcea66490cbc32a9bea0c6"}]},"branch":"refs/heads/master"},"b12112f231e0ffa3090477bff0e84e02e5732cff":{"kind":"REWORK","_number":4,"created":"2019-09-10 18:19:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/672763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/4"}}},"commit":{"parents":[{"commit":"d19fa2a83939f2f5775a8b1e6d9c8ce7031f89af","subject":"Merge \"Blacklist eventlet 0.21.0,0.23.0,0.25.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d19fa2a83939f2f5775a8b1e6d9c8ce7031f89af"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-07-25 15:51:40.000000000","tz":-300},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-10 18:19:36.000000000","tz":120},"subject":"Add update_reservation to floating IP plugin","message":"Add update_reservation to floating IP plugin\n\nThe update reservation functionality was not implemented in the original\nplugin. Consequently, any mixed resource lease containing a floating IP\nresource would fail when the lease was updated.\n\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b12112f231e0ffa3090477bff0e84e02e5732cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b12112f231e0ffa3090477bff0e84e02e5732cff"}]},"branch":"refs/heads/master"},"ad6a995852c7df4affb016c85cc37a8b379c0015":{"kind":"REWORK","_number":5,"created":"2019-09-11 16:00:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/672763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/5"}}},"commit":{"parents":[{"commit":"520005b13dc47bf78db33a526c1947e4b2c0d681","subject":"Add microversion support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/520005b13dc47bf78db33a526c1947e4b2c0d681"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-11 16:00:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-11 16:00:07.000000000","tz":120},"subject":"Add update_reservation to floating IP plugin","message":"Add update_reservation to floating IP plugin\n\nThe update reservation functionality was not implemented in the original\nplugin. Consequently, any mixed resource lease containing a floating IP\nresource would fail when the lease was updated.\n\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ad6a995852c7df4affb016c85cc37a8b379c0015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ad6a995852c7df4affb016c85cc37a8b379c0015"}]},"branch":"refs/heads/master"},"2d58351a2a43edb86515d5c0437a19afdc71f19a":{"kind":"REWORK","_number":6,"created":"2019-09-12 13:56:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/672763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/6"}}},"commit":{"parents":[{"commit":"520005b13dc47bf78db33a526c1947e4b2c0d681","subject":"Add microversion support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/520005b13dc47bf78db33a526c1947e4b2c0d681"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-12 13:56:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-12 13:56:36.000000000","tz":120},"subject":"Add update_reservation to floating IP plugin","message":"Add update_reservation to floating IP plugin\n\nThe update reservation functionality was not implemented in the original\nplugin. Consequently, any mixed resource lease containing a floating IP\nresource would fail when the lease was updated.\n\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2d58351a2a43edb86515d5c0437a19afdc71f19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2d58351a2a43edb86515d5c0437a19afdc71f19a"}]},"branch":"refs/heads/master"},"3137ffc84d7791aa5daa5f12dd2af68386812505":{"kind":"REWORK","_number":7,"created":"2019-09-13 09:58:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/672763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/672763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/672763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/672763/7"}}},"commit":{"parents":[{"commit":"520005b13dc47bf78db33a526c1947e4b2c0d681","subject":"Add microversion support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/520005b13dc47bf78db33a526c1947e4b2c0d681"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-12 13:56:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-13 09:58:32.000000000","tz":120},"subject":"Add update_reservation to floating IP plugin","message":"Add update_reservation to floating IP plugin\n\nThe update reservation functionality was not implemented in the original\nplugin. Consequently, any mixed resource lease containing a floating IP\nresource would fail when the lease was updated.\n\nCo-Authored-By: Jacob Colleran \u003cjakecoll@uchicago.edu\u003e\nChange-Id: I2f55f33ba5af9d9f2b84b716caaa8ed1cfd0db66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3137ffc84d7791aa5daa5f12dd2af68386812505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3137ffc84d7791aa5daa5f12dd2af68386812505"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
