)]}'
{"id":"openstack%2Fblazar~672885","triplet_id":"openstack%2Fblazar~master~I6cf3c51168ea0571501a01ca44ea2134ec281f68","project":"openstack/blazar","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I6cf3c51168ea0571501a01ca44ea2134ec281f68","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-07-26 06:47:15.000000000","updated":"2019-08-22 10:26:44.000000000","submitted":"2019-08-22 10:24:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672885-1566469499046-d74ab91c","meta_rev_id":"6583f6cc908c40231e2c35b7af72148a10f8cf65","_number":672885,"virtual_id_number":672885,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-08-22 10:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":2,"date":"2019-08-22 10:12:12.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-30 09:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":1,"date":"2019-08-22 10:12:12.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 08:42:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-22 10:12:12.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-08-22 10:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3901358c27eca0ac08c8179ee4931850410b196a","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-26 06:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ca3f851801db8d461829b2b25b9f8f4e31a5d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 07:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/85/672885/1/check/build-openstack-api-ref/6fbed99/html/ : SUCCESS in 6m 06s\n- requirements-check http://logs.openstack.org/85/672885/1/check/requirements-check/a95b597/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672885/1/check/openstack-tox-lower-constraints/8f1595e/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/85/672885/1/check/openstack-tox-pep8/fbd2ae1/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/85/672885/1/check/openstack-tox-py27/ec9230e/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/85/672885/1/check/openstack-tox-py36/7790eed/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/85/672885/1/check/openstack-tox-py37/8561153/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/85/672885/1/check/openstack-tox-docs/21b7595/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/85/672885/1/check/build-openstack-releasenotes/11ecb78/html/ : SUCCESS in 5m 09s\n- openstack-tox-pylint http://logs.openstack.org/85/672885/1/check/openstack-tox-pylint/fbd62cc/ : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29ce9c61cdde0e2371f2a7f14bbc9faf7220ec94","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-26 07:43:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"feb08d0b45f95118f55da8a72668a811be139a98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 08:33:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"448872054e2df4f61112bc2eaa525c5d6055ee72","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 08:42:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"70b5a2d53fc3adbe5231699ce82e2a102306c7f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 08:42:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41d155df7ab66caf34f2d66e4c1d70b06e3feed1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 08:58:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/85/672885/2/check/build-openstack-api-ref/9ff0f9c/html/ : SUCCESS in 4m 38s\n- requirements-check http://logs.openstack.org/85/672885/2/check/requirements-check/027fc8c/ : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672885/2/check/openstack-tox-lower-constraints/885c9ed/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/85/672885/2/check/openstack-tox-pep8/72c3a07/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/85/672885/2/check/openstack-tox-py27/03bd048/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/85/672885/2/check/openstack-tox-py36/bcf340f/ : SUCCESS in 6m 36s\n- openstack-tox-py37 http://logs.openstack.org/85/672885/2/check/openstack-tox-py37/7bc107d/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/85/672885/2/check/openstack-tox-docs/860873e/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/85/672885/2/check/build-openstack-releasenotes/a8bfe99/html/ : SUCCESS in 3m 42s\n- openstack-tox-pylint http://logs.openstack.org/85/672885/2/check/openstack-tox-pylint/4c7eda9/ : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81101baf170841bfa834ae88a474615e12520665","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-30 09:01:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1b339306416861501611a08133af88d9d00035ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:15:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/85/672885/3/check/build-openstack-api-ref/d43cd65/html/ : SUCCESS in 4m 42s\n- requirements-check http://logs.openstack.org/85/672885/3/check/requirements-check/c2ad533/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672885/3/check/openstack-tox-lower-constraints/bf39aba/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/85/672885/3/check/openstack-tox-pep8/dca9c72/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/85/672885/3/check/openstack-tox-py27/2dada72/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/672885/3/check/openstack-tox-py36/9a7f77d/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/85/672885/3/check/openstack-tox-py37/fe96a60/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/85/672885/3/check/openstack-tox-docs/23c1457/html/ : SUCCESS in 3m 30s\n- build-openstack-releasenotes http://logs.openstack.org/85/672885/3/check/build-openstack-releasenotes/201fbad/html/ : SUCCESS in 5m 12s\n- openstack-tox-pylint http://logs.openstack.org/85/672885/3/check/openstack-tox-pylint/59ac437/ : SUCCESS in 7m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"681f6d4a93a3f5c59d44d3ddadbe899ed2d113bd","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-08-22 10:12:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"df9d0437b52e747bf5cb3bddf5bc4810b5530730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:12:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"328386c0bcfeae53c834a89e92294d5c0add5325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:24:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf309f57a21544b58c186065cdcf452d : SUCCESS in 3m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1c02e9607f94f3ba31c2467e8e67fc3 : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1714c3376364f20a2a0c8de145ac176 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c276e039974d5ebd0a9a85a184077b : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/502a82a0125b48f5af37d1441e6cde45 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0972452454ac407faadf968729114f0e : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70f01e650cdc45589beb0d779497d04b : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4e163e513f447dbe58b8c0db5dfe45 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e7bcc0e4e494f9e87dc6fd8e4c85328 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"5954bfdc89aa2848de1b11c83a40631e659c4fb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:24:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4e50b9dca322a0e1d8ff80e06503a2b8b5254134","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:26:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b8d5f22af3a0455c86c64d7ce2ee950f : SUCCESS in 1m 18s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94501aa972df4adca4b86fe0d6243511 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c10c0705411e95a4f70af5385945544b2979aa0","revisions":{"5287b2131ad0e8462909f7aae0ca7fcf0da7d385":{"kind":"REWORK","_number":1,"created":"2019-07-26 06:47:15.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/85/672885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/85/672885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/85/672885/1"}}},"commit":{"parents":[{"commit":"125f53bf483349898138a7cb945832bc2efaad5b","subject":"Merge \"Delete specs from blazar repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/125f53bf483349898138a7cb945832bc2efaad5b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 06:46:14.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 06:47:27.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\nDepend-On: https://review.opendev.org/#/c/663060/\n\nChange-Id: I6cf3c51168ea0571501a01ca44ea2134ec281f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5287b2131ad0e8462909f7aae0ca7fcf0da7d385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5287b2131ad0e8462909f7aae0ca7fcf0da7d385"}]},"branch":"refs/heads/master"},"bc6b484f5e7edb715842a141e1ef31f991bbf8a6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-30 08:42:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/672885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/85/672885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/85/672885/2"}}},"commit":{"parents":[{"commit":"125f53bf483349898138a7cb945832bc2efaad5b","subject":"Merge \"Delete specs from blazar repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/125f53bf483349898138a7cb945832bc2efaad5b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 06:46:14.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-30 08:33:46.000000000","tz":0},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: I6cf3c51168ea0571501a01ca44ea2134ec281f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bc6b484f5e7edb715842a141e1ef31f991bbf8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bc6b484f5e7edb715842a141e1ef31f991bbf8a6"}]},"branch":"refs/heads/master"},"7c10c0705411e95a4f70af5385945544b2979aa0":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-30 09:01:26.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/85/672885/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/85/672885/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/672885/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/85/672885/3"}}},"commit":{"parents":[{"commit":"125f53bf483349898138a7cb945832bc2efaad5b","subject":"Merge \"Delete specs from blazar repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/125f53bf483349898138a7cb945832bc2efaad5b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 06:46:14.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-30 09:00:51.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: I6cf3c51168ea0571501a01ca44ea2134ec281f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7c10c0705411e95a4f70af5385945544b2979aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7c10c0705411e95a4f70af5385945544b2979aa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
