)]}'
{"id":"openstack%2Fblazar~679002","triplet_id":"openstack%2Fblazar~master~I7d7c1171b8501e389c5017856bd2173ec6a56cab","project":"openstack/blazar","branch":"master","topic":"addlink","hashtags":[],"change_id":"I7d7c1171b8501e389c5017856bd2173ec6a56cab","subject":"Add blazar-specs link to README.rst","status":"MERGED","created":"2019-08-28 06:48:52.000000000","updated":"2019-09-08 03:24:01.000000000","submitted":"2019-09-08 03:23:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679002-1567912990023-72eb98b0","meta_rev_id":"372aa2541123758e1e944469151e8b04681bf37e","_number":679002,"virtual_id_number":679002,"owner":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-08 03:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":2,"date":"2019-09-08 03:08:28.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-05 14:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":1,"date":"2019-09-08 03:08:28.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 05:52:23.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-08-29 09:24:02.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2019-09-05 14:20:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-08 03:08:28.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-08 03:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3404f1bb6009d4a5d56fb7527b13201ccabbc3e1","author":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"date":"2019-08-28 06:48:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d52a72056541f2782aeb2e7a13e2d61f431fd56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:12:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32fd1f286dc9490c8c68ec443dd297b2 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39f9d2b420a46ffb042af54b6517899 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f097ada1c64938b5beb9411921128f : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ae433e830d21b0f30a0dca6c4008e65722b0c526","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-28 07:22:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e2d7585bdf114ce9c29e47543730fdf72c283430","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-28 07:22:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9aee3716a3a6558168aee50886b123a243292a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:32:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/537afbda150740ee90c79634bc48bb88 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9570da38a7405fb698de799efbce97 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41df5c8721c43a3a3011b003a22ac00 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6b58d1dfeb49d16f2fc917c2245f299a59db7a36","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-08-29 05:52:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b30a668db43349d0657da40981236703848bc31","author":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"date":"2019-08-29 09:20:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"84ac03beab28cee8e4aff5040a5785f40e0dd34d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 09:36:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a606c2d6d4b4e22951389cf32cae594 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851bb3ff550c47dbba7e005f63048c46 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3db45d02c56476492a9d8f361e0b7b3 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b1ae762d41e1ca99746f4785332766c03dd2b89f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-03 09:00:51.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c3345b727a20b8cbb3050725e8f723bdcd188071","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-04 02:08:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"177b69c5e1091ae6ed8e2436d6751cf15d83ed1f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-05 14:20:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b41348f3a975ffe6f7b314e99682ab4a69f62760","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-05 14:20:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fab52cd014a455b6bb1192dcab8cd5e0a8cc2b3f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-05 14:20:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9b1454ad6f9ab86d11cfda7bfe133318e921224b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 14:31:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43b7d5525a774485957cec17b9e5a377 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf3ed78e1494dc184bdcd972fd49d26 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a284c1431b1140b5b5c0c5cbcec5bb42 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":5},{"id":"2aed67cb1990861634d50b639db3f9e2e40dca0c","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-08 03:08:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"abc39f249332466e63f3fbfaf97ed6320e56d9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 03:08:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"308702123ec43a4d056608dd3c1c1e0bd5680d44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 03:23:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5619201ef264923b1d0b51d32a7fa46 : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3031e2033f2b4eeea8bf2b8302807955 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/352ab59fe12445f5bc3efb0f5e32582a : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":5},{"id":"86c00ec7edc7875a564f363bcc927d98a82cf528","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 03:23:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fe68341326896ba70fc24ec26e86d1a1623536cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 03:24:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac40a824648444ab5a4cf47a5fc4496 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9bcb057f4755911b03bf7f8342ea5cd3b2741f79","revisions":{"e9d4bd946b884a99a905f027b53638cde8273d24":{"kind":"REWORK","_number":1,"created":"2019-08-28 06:48:52.000000000","uploader":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"ref":"refs/changes/02/679002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/02/679002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/02/679002/1"}}},"commit":{"parents":[{"commit":"8212d46d3e55422a3b108399006ca87109326042","subject":"Merge \"Fix placement operations in multi-region deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8212d46d3e55422a3b108399006ca87109326042"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"committer":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"subject":"Add blazar-spec link to readme.rst","message":"Add blazar-spec link to readme.rst\n\nChange-Id: I7d7c1171b8501e389c5017856bd2173ec6a56cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e9d4bd946b884a99a905f027b53638cde8273d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e9d4bd946b884a99a905f027b53638cde8273d24"}]},"branch":"refs/heads/master"},"b59d9f915e1dcd30d7e0e7f9148a06fe93bc7a39":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-28 07:22:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/679002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/02/679002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/02/679002/2"}}},"commit":{"parents":[{"commit":"8212d46d3e55422a3b108399006ca87109326042","subject":"Merge \"Fix placement operations in multi-region deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8212d46d3e55422a3b108399006ca87109326042"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-28 07:22:25.000000000","tz":0},"subject":"Add blazar-specs link to README.rst","message":"Add blazar-specs link to README.rst\n\nChange-Id: I7d7c1171b8501e389c5017856bd2173ec6a56cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b59d9f915e1dcd30d7e0e7f9148a06fe93bc7a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b59d9f915e1dcd30d7e0e7f9148a06fe93bc7a39"}]},"branch":"refs/heads/master"},"93e1f7e57f995f97a14657c64983f1c162969a67":{"kind":"REWORK","_number":3,"created":"2019-08-29 09:20:46.000000000","uploader":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"ref":"refs/changes/02/679002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/02/679002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/02/679002/3"}}},"commit":{"parents":[{"commit":"8212d46d3e55422a3b108399006ca87109326042","subject":"Merge \"Fix placement operations in multi-region deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8212d46d3e55422a3b108399006ca87109326042"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"committer":{"name":"Shuo Liu","email":"liushuobj@inspur.com","date":"2019-08-29 09:20:31.000000000","tz":0},"subject":"Add blazar-specs link to README.rst","message":"Add blazar-specs link to README.rst\n\nChange-Id: I7d7c1171b8501e389c5017856bd2173ec6a56cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/93e1f7e57f995f97a14657c64983f1c162969a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/93e1f7e57f995f97a14657c64983f1c162969a67"}]},"branch":"refs/heads/master"},"0726a0725e3c1383daabaa2be33b4befd6dd1b00":{"kind":"REWORK","_number":4,"created":"2019-09-05 14:20:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/679002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/02/679002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/02/679002/4"}}},"commit":{"parents":[{"commit":"8212d46d3e55422a3b108399006ca87109326042","subject":"Merge \"Fix placement operations in multi-region deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8212d46d3e55422a3b108399006ca87109326042"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-05 14:20:39.000000000","tz":0},"subject":"Add blazar-specs link to README.rst","message":"Add blazar-specs link to README.rst\n\nChange-Id: I7d7c1171b8501e389c5017856bd2173ec6a56cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0726a0725e3c1383daabaa2be33b4befd6dd1b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0726a0725e3c1383daabaa2be33b4befd6dd1b00"}]},"branch":"refs/heads/master"},"9bcb057f4755911b03bf7f8342ea5cd3b2741f79":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-05 14:20:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/679002/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/02/679002/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/02/679002/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/02/679002/5"}}},"commit":{"parents":[{"commit":"fedc28905907d4056b513a09645b4b5bf922d8f9","subject":"Fix gate issue of placement unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fedc28905907d4056b513a09645b4b5bf922d8f9"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-28 06:34:22.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-05 14:20:47.000000000","tz":0},"subject":"Add blazar-specs link to README.rst","message":"Add blazar-specs link to README.rst\n\nChange-Id: I7d7c1171b8501e389c5017856bd2173ec6a56cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9bcb057f4755911b03bf7f8342ea5cd3b2741f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9bcb057f4755911b03bf7f8342ea5cd3b2741f79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
