)]}'
{"id":"openstack%2Fblazar~682416","triplet_id":"openstack%2Fblazar~master~I6820b56d0389ec285789ba38a8f82eac93595afc","project":"openstack/blazar","branch":"master","topic":"floatingip-api-fix","hashtags":[],"change_id":"I6820b56d0389ec285789ba38a8f82eac93595afc","subject":"Fix errors when accessing floating IP API methods","status":"MERGED","created":"2019-09-16 16:56:20.000000000","updated":"2019-09-24 10:23:24.000000000","submitted":"2019-09-24 10:21:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682416-1569320492078-91403195","meta_rev_id":"383aedd600be377650872a4a15f133c0f09f589e","_number":682416,"virtual_id_number":682416,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 10:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":2,"date":"2019-09-24 09:28:13.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":2,"date":"2019-09-19 10:25:48.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-09-17 05:46:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},{"value":2,"date":"2019-09-24 09:25:58.000000000","_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":1,"date":"2019-09-24 09:28:13.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-19 09:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 05:46:02.000000000","updated_by":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"reviewer":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"state":"REVIEWER"},{"updated":"2019-09-19 10:25:48.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-24 09:25:58.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-09-24 09:28:13.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2019-09-24 10:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"783697624294a559a2b90af4d60feb4a9e984396","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-16 16:56:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0364e3d8243fe28584075f4d5aea053caf146431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:57:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/def31e451f7845cbba0b62dde8f1f610 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023a09ba70b54075aa4fbb1c39c6a865 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bff804b830d246e8bab6cff4f7c7ff8a : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58af0ba2810d46f3bbe44620fe60c3ac : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c3cc281f91544d088be11499ac685d3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db2e5646c2540e985fa00b2a4b40801 : SUCCESS in 4m 01s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/f39056260174488a91eb33122beff811 : SUCCESS in 51m 58s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/65cec5360b3c465aae2424378382c525 : SUCCESS in 49m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1aa78f4a301a47d29696b3a5db9a8467 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fbcc56b112fe1e1dd8d49c09029e32653b21c86","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-09-17 05:46:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66d52c6a29e884c922aa73180879cc0fa4785e28","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-18 17:59:01.000000000","message":"Patch Set 1:\n\nMasahito, Tetsuro, Bertrand: please review this important patch.","accounts_in_message":[],"_revision_number":1},{"id":"28c291b74692d94588fa56614580f8bda77a9862","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-19 09:40:32.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"b41542af305a7cc78fcd2c1709a556985c96cef9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-19 10:25:48.000000000","message":"Patch Set 1: Code-Review+2\n\nGood catch, thanks. Can we have the tests under blazar/blazar/tests/api/v1 as well as other api points like leases and oshosts?","accounts_in_message":[],"_revision_number":1},{"id":"949f2166ec45a10117afe7112fd3437df3b974ff","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-09-24 09:25:58.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. sorry slow response.","accounts_in_message":[],"_revision_number":1},{"id":"d647e1547863184be6b957b22afe97e9180d16b1","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2019-09-24 09:28:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"07c89873f8e7b6baffa80469ab0f6e946e5a3030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:28:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e04d19fcab3475518dd824a9bb5cadf34c0ed9d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:21:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1acb1908abcb4a96974a91b3d838cbd1 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2bdf8c4f744ef8a7edcb2b797d02cc : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f4510b35e4f486dbf5680f27bbfa77e : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4e22f34feb049f9ac1fac617941d892 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c13417a68714a00b0e2e7c2cd9c08d6 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ab806a81154e25bd43e4dc9637b295 : SUCCESS in 4m 28s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/e5be48ccc06b43d3ac7ddd17b7883348 : SUCCESS in 46m 35s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/62e59c21083640d3823dd381bdd51d29 : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5d5040e5a166f54091120ba6be911d8d6465dfea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:21:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59342dac0d3c520e523b0f2df7a1dc961eac37f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:23:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab81355dfde2441da164bcb9f7944153 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a775020bcf273550e6202de3a1facf04cc51b56","revisions":{"0a775020bcf273550e6202de3a1facf04cc51b56":{"kind":"REWORK","_number":1,"created":"2019-09-16 16:56:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/682416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/16/682416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/16/682416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/16/682416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/16/682416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/16/682416/1"}}},"commit":{"parents":[{"commit":"d4c6ecf885ea0c70838c3f56132ca3ba6ebf3137","subject":"Merge \"Add update_reservation to floating IP plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d4c6ecf885ea0c70838c3f56132ca3ba6ebf3137"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-16 16:55:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-16 16:55:49.000000000","tz":120},"subject":"Fix errors when accessing floating IP API methods","message":"Fix errors when accessing floating IP API methods\n\nCommit 520005b13dc47bf78db33a526c1947e4b2c0d681 which recently added\nmicroversion support did not update floating IP code to account for the\nextra parameter passed to all API methods. This resulted in errors like:\n\n    File \"/opt/stack/blazar/blazar/api/v1/utils.py\", line 90, in handler\n    return func(flask.request, **kwargs)\n    TypeError: floatingips_list() takes no arguments (1 given)\n\nChange-Id: I6820b56d0389ec285789ba38a8f82eac93595afc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0a775020bcf273550e6202de3a1facf04cc51b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0a775020bcf273550e6202de3a1facf04cc51b56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
