)]}'
{"id":"openstack%2Fblazar~683372","triplet_id":"openstack%2Fblazar~master~I96d8b2fa9475c097e9eda6ecfc7ebb5a9f471567","project":"openstack/blazar","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I96d8b2fa9475c097e9eda6ecfc7ebb5a9f471567","subject":"Fix empty sections in pdf document","status":"MERGED","created":"2019-09-20 12:09:40.000000000","updated":"2019-09-27 12:49:47.000000000","submitted":"2019-09-27 12:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683372-1569588465046-02882d20","meta_rev_id":"2daa4f10a54aa5a41c4f24e55c9349820ac296b9","_number":683372,"virtual_id_number":683372,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 12:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-26 13:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-27 11:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-26 13:31:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 11:54:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-27 12:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a44be7bb5448cb0124f43317a427b30b69273b9a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-20 12:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb2d980ba3d5d737a3778d49d0152680574d0b07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:58:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e613979b441d4f73b07d1823d117e2be : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91143fbb168940df92abe1b9d1f3e7ed : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f158b109fcd44debc156a2c7d3739d7 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14fbd301fd9d46ff9f65f6ab5e4e633d : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8afa6776c4e4613af9fdfe2daf145fa : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baeb3ebefbe049d6963b6493551e4909 : SUCCESS in 6m 30s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/c30754f145e443ae8c48ae1bbadecb8d : SUCCESS in 37m 33s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/fcfce53d890d43e3bfdd79987153000f : SUCCESS in 41m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5336229fccbe4bf1aea5ce7dc170029f : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"484e6520c7d546dc9bf1e05c8fe5b5ad94ba3ead","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-23 07:37:31.000000000","message":"Topic changed from pdf-docs to build-pdf-docs","accounts_in_message":[],"_revision_number":1},{"id":"8d167597445580e8e849f4ddda3922f92fd0c6aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-23 07:44:36.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you Tetsuro!","accounts_in_message":[],"_revision_number":1},{"id":"63643dabe24e808d6130ca30f4d3aba35bee3cac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-24 09:38:41.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"182907567cb46669161cd378bec39b9ca6a490c3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-26 11:34:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e1ed786d15d11f09006e7a78ff1e232292f93d3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:32:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5640b2e9302b493b826de8d365007323 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b999018094d849f996864923d74b7306 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b90e517f75e34855902655be0410897c : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e97aa160cef47759f3387d1c2e39395 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30c5cc413e80405780aa2bade70191cf : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8b4c67d1184efabe6efdc948e66900 : SUCCESS in 9m 53s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/5be8b17eb398470fabfb699a595bb1e6 : SUCCESS in 50m 52s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/7e260ed6d0f747d7ac72bf519b47f9b5 : SUCCESS in 44m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2097734b17dd436e9044e1e3a523d0a1 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a80b7ff2beba7a104f117dd90818e945b3924d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-26 13:31:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e5079cebf56d0f80e2ba8fc722360eb5bdb3de45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:22:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ff470abe6df44feaaa009f4f1392646 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/342b2649ec9d4959aa7b89f0907c1f6c : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4d5febf95ce48ed92f96e5bf177343c : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512a7cc15fba4d018e27887bf8382b9d : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/565bc719da6145e7b62a91b399acaad3 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806cc8ee828149038357595cbc1474ef : SUCCESS in 7m 19s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/e9706c41c2064fe08a99fe15d0dbbeb7 : SUCCESS in 42m 59s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/d46e53aa6acb4a32b863349961c512aa : SUCCESS in 39m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0fd6dada147342bcbef71c998dd1e744 : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"882d2ff1a57e908be50e0acfb33740ffaa0be09a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-27 09:45:43.000000000","message":"Patch Set 3:\n\nMerging with a single +2 as this is required for completing Train community goals.","accounts_in_message":[],"_revision_number":3},{"id":"bd9640393e034b76c75e0a2cc7f1702f0c642030","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-27 11:54:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"21c3e091539364660b84e218deff7d2b8148eba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 11:54:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"710767b3c8a5f155b0dc9fde60ba88cb023752ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:47:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f055da5589af47a58f76c21a633ed885 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c343575994b14bfb850e51d1686d7d70 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41214bb6362b4a5386c26b97662db770 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb15410d9b3f4f39b8d369324c048251 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50346f0adabc44b185e6cd879fb3148d : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf9a076d1fd43a1a99a9f48d1a6eae6 : SUCCESS in 6m 48s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/a5c8cb4819fa477294612ce9c04fd6ea : SUCCESS in 42m 07s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/ce9700e6d26e4806a28473d44897a863 : SUCCESS in 37m 59s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/928d0bb9ab3d47e38186d7362e0c8da2 : SUCCESS in 43m 08s","accounts_in_message":[],"_revision_number":3},{"id":"1b6254dd19e32147c8aeb9825453fc68be53cdb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:47:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2daa4f10a54aa5a41c4f24e55c9349820ac296b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:49:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991002db6d9340e39532946a4fb350f8 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b0ed63db7dd82dee5c69500c6718a62f7873d0e8","revisions":{"9a69dee07fec547829a46bcd992149a95839985a":{"kind":"REWORK","_number":1,"created":"2019-09-20 12:09:40.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/72/683372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/72/683372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/72/683372/1"}}},"commit":{"parents":[{"commit":"d1917d14ea5b909965d45f2d755f7b3c709597bc","subject":"Build pdf doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d1917d14ea5b909965d45f2d755f7b3c709597bc"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-20 12:03:51.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-20 12:08:54.000000000","tz":0},"subject":"Fix empty sections in pdf document","message":"Fix empty sections in pdf document\n\nThe following sections were empty in the pdf document.\n\n  3. Installation Guide\n  4. Configuration Reference\n  5. CLI Reference\n  6. API Reference\n\nThis was because it is already referred in the second section.\n\n  2. Admin Guide\n\nThis patch fixes the empty sections by moving the admin guide after\nthose sections.\n\nChange-Id: I96d8b2fa9475c097e9eda6ecfc7ebb5a9f471567\nStory: 2006073\nTask: 34795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9a69dee07fec547829a46bcd992149a95839985a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9a69dee07fec547829a46bcd992149a95839985a"}]},"branch":"refs/heads/master"},"4d0472979543b3f53ad6d714a9a305f739143612":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-26 11:34:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/683372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/72/683372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/72/683372/2"}}},"commit":{"parents":[{"commit":"7e8b85d7fcfc14573a61945e0cd64177b1acbc6f","subject":"Build pdf doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7e8b85d7fcfc14573a61945e0cd64177b1acbc6f"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-20 12:03:51.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-26 11:33:43.000000000","tz":120},"subject":"Fix empty sections in pdf document","message":"Fix empty sections in pdf document\n\nThe following sections were empty in the pdf document.\n\n  3. Installation Guide\n  4. Configuration Reference\n  5. CLI Reference\n  6. API Reference\n\nThis was because it is already referred in the second section.\n\n  2. Admin Guide\n\nThis patch fixes the empty sections by moving the admin guide after\nthose sections.\n\nChange-Id: I96d8b2fa9475c097e9eda6ecfc7ebb5a9f471567\nStory: 2006073\nTask: 34795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4d0472979543b3f53ad6d714a9a305f739143612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4d0472979543b3f53ad6d714a9a305f739143612"}]},"branch":"refs/heads/master"},"b0ed63db7dd82dee5c69500c6718a62f7873d0e8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-26 13:31:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/683372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/72/683372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/683372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/72/683372/3"}}},"commit":{"parents":[{"commit":"06d102c655ca4d8fb26c1aac03eed0340a5d3211","subject":"Build pdf doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/06d102c655ca4d8fb26c1aac03eed0340a5d3211"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-20 12:03:51.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-26 13:30:56.000000000","tz":120},"subject":"Fix empty sections in pdf document","message":"Fix empty sections in pdf document\n\nThe following sections were empty in the pdf document.\n\n  3. Installation Guide\n  4. Configuration Reference\n  5. CLI Reference\n  6. API Reference\n\nThis was because it is already referred in the second section.\n\n  2. Admin Guide\n\nThis patch fixes the empty sections by moving the admin guide after\nthose sections.\n\nChange-Id: I96d8b2fa9475c097e9eda6ecfc7ebb5a9f471567\nStory: 2006073\nTask: 34795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0ed63db7dd82dee5c69500c6718a62f7873d0e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0ed63db7dd82dee5c69500c6718a62f7873d0e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
