)]}'
{"id":"openstack%2Fblazar~691545","triplet_id":"openstack%2Fblazar~master~I69bd82f8a1156f2f2eb636ea96c439c0645cc102","project":"openstack/blazar","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I69bd82f8a1156f2f2eb636ea96c439c0645cc102","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-10-28 01:57:15.000000000","updated":"2020-01-07 13:43:19.000000000","submitted":"2020-01-07 13:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691545-1578404417093-0e1c537b","meta_rev_id":"3e3975a1e2a816f47723599a573a4b214901d6ff","_number":691545,"virtual_id_number":691545,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 13:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-13 00:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-12-10 09:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-01-07 12:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-01-07 08:53:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 13:44:55.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-12-13 00:14:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-07 12:45:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-07 13:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"247e75f178a746d205fe7bb0e1a491bbfb89af22","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-10-28 01:57:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77d6bcec7b3b14022c441a4875493b8a862d8a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 02:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4ca8e94f982144d187d5f3fe11b3efa6 : SUCCESS in 3m 12s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f76332084a24e988273553bc6d2e66b : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce9dd03293aa432db5a55058a5020b8d : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61ff1523f064546a030574325bc1c41 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/706ad6b2ee7b4b4bb48ed892f18ded34 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f73c3c8c32514caaac867aa6f77615a1 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da374d88e52443cebaaeaab878c2c877 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe3f79d002340408c82cc8d8f100882 : SUCCESS in 4m 36s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/9a32ae92cf6c45388cf28bea1d284616 : SUCCESS in 43m 13s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/7b554804330347d7805d37fc1ad51d37 : SUCCESS in 38m 23s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f4de745065cd4379b5e350f9f9d2635b : SUCCESS in 38m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4797ae3f67be418b9c04de4f09f66dfc : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c184465faed20f036fd41a1bd0d7301794676b4","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 13:44:55.000000000","message":"Patch Set 1: Code-Review-1\n\nwe need to do a lot more on this and as per discussed schedule. IGhanshyam Mannam composing the community-wide goal  -https://review.opendev.org/#/c/691178/\n\nLet\u0027s be on sync on what all to do and when via goal.","accounts_in_message":[],"_revision_number":1},{"id":"98b1dd470d1a987219e6fb196978348e69128aae","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 01:53:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1500da717d222a058c5f32477f57d4baa9f5246c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 01:56:41.000000000","message":"Patch Set 2:\n\nI made blazar-devstack-dsvm running on py3 and renamed blazar-devstack-dsvm-py35 -\u003e blazar-devstack-dsvm-","accounts_in_message":[],"_revision_number":2},{"id":"6eb34ea35c16a5b471b08e27b364cf167fab5dc3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 02:21:05.000000000","message":"Topic changed from drop-python2 to drop-py27-support","accounts_in_message":[],"_revision_number":2},{"id":"7e0646247d63579ad52a7b08a2faf81d23964070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 02:33:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e82de1c472a14a59942fc0c1466a97eb : SUCCESS in 2m 38s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/383b94cfd03149a3bea9e21c2984a1d0 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6250faf19fea44d69190ba673d1cd470 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9df71e5090845d89ec7df50cb1cf7ff : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94463c1e67de4157b67a8cfbc6618beb : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bbc4439068347ccb2f4db28f789209b : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3c0b4dbcc75414f957a1a816667e451 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1047a682d84543a7ac9237988cdb87ea : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f50fc9fb3784971bb3c41c9356e97f3 : SUCCESS in 3m 57s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/e8a1309d7137469fbcade26873754e69 : SUCCESS in 37m 45s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/972a6f33e53d422ca25f0d723b3742a0 : SUCCESS in 38m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/45adc64b9170480092b1f77db4b399ca : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09b3e319c70e5370a8da0cc32a6c39c0ad9e25a9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-05 20:51:31.000000000","message":"Patch Set 2: Code-Review+1\n\nthis is good to go now.","accounts_in_message":[],"_revision_number":2},{"id":"2d0a7ece931365fef6878cf85c92f45728ee35a8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 04:14:49.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9fa480c955c3a4c585e6e8d107dce5fbd894af84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:59:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9cd1f620728c4b29aec833f0cfe5f1ad : SUCCESS in 2m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af4a7dcb951144d28cae7515604928d2 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49f34ca42b8b4fb98215a397b90b7871 : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9b14693c3c5405686900bebdc7d56bd : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7427830ccf5449c89aaa6584dd9675bc : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8152e108e01849a9bce5ed4defa29cbe : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/148f8a92ab2c4122a93c788214176765 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c07659cdc84734a099be711ebce424 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/846ff892767b4b62ae34416924800137 : SUCCESS in 3m 59s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/d275a10800a54a04bf9e7046712e4ad6 : SUCCESS in 37m 10s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/a86d5230bab4447e83a4d57aa1194a6a : SUCCESS in 40m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e95352d2324341f79d1fac0412ad07ee : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61bee8290ddcac7fb5c50b4c15e8aa16731d3de2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-10 09:14:21.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"7f4cdc2a41bd99e71c32e8d6d2c5b31f4bfe1bef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-10 09:16:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"648881cf58a0b4339587a2238d2cefc72b328784","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-10 09:20:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fea24bcf59fe26b1c92e02927b8503210f7d424c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-10 09:20:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e46def35c87edc687413c3f99479b485fe52c08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:03:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99d577b300fe484d970b1e5da981fd8f : SUCCESS in 2m 51s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/39a9f4f83ee1490aaf7c439a6f27e36b : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78389d28bb914982bdc7734aa77031dd : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c5a38ba61cd4745b2c50690635d5351 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8acbc675134739a3a8f331eb79b209 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/235a9c1c7b654dccbeccccbb6f13b589 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57b906b820354d08bb004c4a37188e7a : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae674a3dbc4b44f88e501ebabd00bf6c : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3433a2e2c8491ba11fbaa200d9e9d9 : SUCCESS in 3m 38s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/71b073dfa728493caf53e1eeb3535248 : SUCCESS in 38m 58s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/54e6150cba454f88986ac2ec0161c523 : SUCCESS in 37m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd8672de4745467c831c5fc61873b196 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2656383fb0e7342e8bf22b402796a2bebf1f37c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-13 00:14:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c1cd5ad44ddd9dab8c93f1e385a2c9981e5dfd6e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-07 08:53:15.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"2e864eded393dd2117e2089a12fcb8f0d6e51acf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-07 12:45:49.000000000","message":"Patch Set 5: Workflow+1\n\nApproving with a single +2 due to lack of core reviewers. Tetsuro\u0027s +2 is implied since he is the author of the change.","accounts_in_message":[],"_revision_number":5},{"id":"3e8e9cd3e500343a5a49a4836e9eb16015b06bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 12:46:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"30e9036cc42c15004aa7fe26fda8e3a247c9b266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 13:40:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dbd81747b15416ca8114862d521487a : SUCCESS in 2m 59s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b2d55c5ee7904578bf31ab5df38530bf : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fca9ca56c1b64b598dad8a5b73ca19c6 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9b837abc9d438c8f2b8193ee1b81df : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c15f72b2fc034ddf8b8280243e6b901c : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e78b218f3f3147d3882359588d844e56 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a0cc3fd18b4c5ea26549f4693367b9 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53fd16454cf54605a3f50cf76773fb1e : SUCCESS in 5m 00s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/6299b6760545456a9f41d1f45b77bac2 : SUCCESS in 47m 36s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/4da5c224baf54911b3d43af8e755bda6 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":5},{"id":"fd76e65ba11acfa67cd09b72e2e4b2fd7efb67d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 13:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3e3975a1e2a816f47723599a573a4b214901d6ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 13:43:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ed9d728a4a4c4e2687a93c9331745bd7 : SUCCESS in 1m 28s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d1e2827eb84e2e8356b6166f3bf883 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33988377967c461b8716c2eec03dbce9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"352893e78cfc78b11384f80056b570433a7727ee","revisions":{"d1be3c2b29dc450b49109b384f9af5752106b0b4":{"kind":"REWORK","_number":1,"created":"2019-10-28 01:57:15.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/45/691545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/45/691545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/45/691545/1"}}},"commit":{"parents":[{"commit":"651f52be5998ec7b82c65975c854a562890ae602","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/651f52be5998ec7b82c65975c854a562890ae602"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:53:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:56:44.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stop testing with python2 and issue the release note.\n\nChange-Id: I69bd82f8a1156f2f2eb636ea96c439c0645cc102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d1be3c2b29dc450b49109b384f9af5752106b0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d1be3c2b29dc450b49109b384f9af5752106b0b4"}]},"branch":"refs/heads/master"},"def44541837c7e18a997ef458baaf447a3be878f":{"kind":"REWORK","_number":2,"created":"2019-11-15 01:53:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/691545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/45/691545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/45/691545/2"}}},"commit":{"parents":[{"commit":"b76daf1ff60c2ee168d073029b87d2932167084a","subject":"Add max-complexity in tox pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b76daf1ff60c2ee168d073029b87d2932167084a"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:53:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 01:51:54.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nBlazar is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCo-Authored-By: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\n\nChange-Id: I69bd82f8a1156f2f2eb636ea96c439c0645cc102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/def44541837c7e18a997ef458baaf447a3be878f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/def44541837c7e18a997ef458baaf447a3be878f"}]},"branch":"refs/heads/master"},"1ebd3a18247c5e7fea1550325be18cadb9e2207b":{"kind":"REWORK","_number":3,"created":"2019-12-06 04:14:49.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/45/691545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/45/691545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/45/691545/3"}}},"commit":{"parents":[{"commit":"b76daf1ff60c2ee168d073029b87d2932167084a","subject":"Add max-complexity in tox pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b76daf1ff60c2ee168d073029b87d2932167084a"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:53:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-12-06 04:14:47.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nBlazar is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCo-Authored-By: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\n\nChange-Id: I69bd82f8a1156f2f2eb636ea96c439c0645cc102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1ebd3a18247c5e7fea1550325be18cadb9e2207b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1ebd3a18247c5e7fea1550325be18cadb9e2207b"}]},"branch":"refs/heads/master"},"3c97412f5fcf7b97ce538eccd9a002480addc4ce":{"kind":"REWORK","_number":4,"created":"2019-12-10 09:16:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/691545/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/45/691545/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/45/691545/4"}}},"commit":{"parents":[{"commit":"b76daf1ff60c2ee168d073029b87d2932167084a","subject":"Add max-complexity in tox pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b76daf1ff60c2ee168d073029b87d2932167084a"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:53:17.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-10 09:16:05.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nBlazar is ready with python 3 and ok to drop python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Community-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCo-Authored-By: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\n\nChange-Id: I69bd82f8a1156f2f2eb636ea96c439c0645cc102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3c97412f5fcf7b97ce538eccd9a002480addc4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3c97412f5fcf7b97ce538eccd9a002480addc4ce"}]},"branch":"refs/heads/master"},"352893e78cfc78b11384f80056b570433a7727ee":{"kind":"REWORK","_number":5,"created":"2019-12-10 09:20:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/691545/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/45/691545/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/45/691545/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/45/691545/5"}}},"commit":{"parents":[{"commit":"9cef00c2ad795d7d6c334451ae7a3fa70f39eaa9","subject":"Merge \"Update Gerrit links following the move to opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9cef00c2ad795d7d6c334451ae7a3fa70f39eaa9"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 01:53:17.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-10 09:20:00.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nBlazar is ready with python 3 and ok to drop python 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Community-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCo-Authored-By: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\n\nChange-Id: I69bd82f8a1156f2f2eb636ea96c439c0645cc102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/352893e78cfc78b11384f80056b570433a7727ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/352893e78cfc78b11384f80056b570433a7727ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
