)]}'
{"id":"openstack%2Fblazar~719910","triplet_id":"openstack%2Fblazar~master~Iee0c8b481ecbe0187eb9acc13bf41c6363bd95ed","project":"openstack/blazar","branch":"master","topic":"fix-docs-publishing","hashtags":[],"change_id":"Iee0c8b481ecbe0187eb9acc13bf41c6363bd95ed","subject":"Fix docs publishing","status":"MERGED","created":"2020-04-14 10:52:14.000000000","updated":"2020-04-14 12:56:39.000000000","submitted":"2020-04-14 12:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719910-1586868516354-a88869dc","meta_rev_id":"22f8ed0bcb6c2844cd3d5ab0da337f7f9619b774","_number":719910,"virtual_id_number":719910,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 12:48:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-14 12:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-14 12:31:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 12:31:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-14 12:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e26a0582157bda86fc2ee7fbd470ccbcf87875d9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-14 10:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f4a35b353ac16b1b96132ca2dfd26a9445d730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc810e2e63a749c996145118fd88b080 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53d15045d5aa4064b5d2af50b20351f4 : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fcb536bce8241ce9354af3c7f1f488d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d9c58fb4f24a6e96b143ede04654ff : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08ed5d943a30422687de7aaefa350e33 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97eb2c368bc24175b88e1fa49999ae57 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51fc177fb33a41c8a4cd7197633bb88d : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d34b12bc7a24bea8a4f3e2d9a612c33 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c448eb21b16c4cd984fc9ff3f057d049 : SUCCESS in 6m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6218a68eacf24cf08cf02a3812c3c1c6 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7717cbca84e7e6d1fcd6c746dfa4f6e4a6b09dae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-14 12:30:46.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6f39896397a5a44ad18be19bb4d05ba6ed581ee4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-14 12:31:58.000000000","message":"Patch Set 1: Workflow+1 -Review-Priority\n\nApproving since it fixes doc build breakage. Verified by looking at artifacts generated by openstack-tox-docs.","accounts_in_message":[],"_revision_number":1},{"id":"11bd41e8f3ab2caf4b4016a356c5d11795618bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82b0742258810d2249c7dc400c0dc91e74203d91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:48:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ca190abb4fbbe47f24c6dff35a2410ec62b2eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:48:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/723f8c17082043adb67a1a674d7d5c80 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e1588786434c6fa24fc05318710e8d : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c1092076cc47c49c7915c3d3f1ca8a : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/665c6460175842729e1ba856407fe57e : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b379319f17534728b4de11381475afe6 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7dc3e3b8c94169b06a2182fde38c36 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ffe71cea3aa43799886f73318b8ddaf : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"56944357940c2ecebb6375ecb0e4e199c7f51dff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:56:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1a2e9d7d5dd24466b4fb6a78631c949b : SUCCESS in 1m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a2e8719045459faebf563dede6f8a1 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31021fa37d0b41c49d63c77cb15f8161 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"738f22c4a9dc2ea30757ee42f12abebe900ea68b","revisions":{"738f22c4a9dc2ea30757ee42f12abebe900ea68b":{"kind":"REWORK","_number":1,"created":"2020-04-14 10:52:14.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/719910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/10/719910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/719910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/719910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/719910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/10/719910/1"}}},"commit":{"parents":[{"commit":"a98989b2ef3a7473473a92d762a22dfe6d704ada","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a98989b2ef3a7473473a92d762a22dfe6d704ada"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-14 10:51:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-14 10:51:36.000000000","tz":120},"subject":"Fix docs publishing","message":"Fix docs publishing\n\nDocs need to end in doc/build/html for proper publishing, fix\nchange I8c81e0c1944c239e460befbf886993270902ee04.\n\nChange-Id: Iee0c8b481ecbe0187eb9acc13bf41c6363bd95ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/738f22c4a9dc2ea30757ee42f12abebe900ea68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/738f22c4a9dc2ea30757ee42f12abebe900ea68b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
