)]}'
{"id":"openstack%2Fblazar~720548","triplet_id":"openstack%2Fblazar~master~Id3531226dc4d10b46d26820f3f47e2aefb3089a6","project":"openstack/blazar","branch":"master","topic":"ironic-node-polling","hashtags":[],"change_id":"Id3531226dc4d10b46d26820f3f47e2aefb3089a6","subject":"Prevent reserving ironic nodes in error states","status":"NEW","created":"2020-04-16 16:30:46.000000000","updated":"2020-04-16 21:10:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe7061f5a6122f610a7c08e2ec9046de2ce5f54e","_number":720548,"virtual_id_number":720548,"owner":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-16 20:54:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-16 21:10:35.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 20:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-16 21:10:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"8b5c122e65cdb50bd7efc8b106c51dc92288d8c3","author":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"date":"2020-04-16 16:30:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5147454b96c037463234520f441d385be53ebc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:12:25.000000000","message":"Patch Set 1: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c65d0bd75bf54aefb5201c1eea7ca044 : POST_FAILURE in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f28922da4be4768bea2f3d0ce6258cb : POST_FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3689ed071fc3453f89397aef411f92a6 : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5791f524abc4c1a988dcd968e93c985 : POST_FAILURE in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed07d1b3b6ce44b788e24b91a1fdf0f9 : POST_FAILURE in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86305fdb290648f7a8e882660dc617ed : POST_FAILURE in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bda30e06e84c45845e7c271b3d6115 : FAILURE in 6m 10s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/14efbb7dc83a479fa71e2312c0772fe2 : FAILURE in 28m 10s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0789da83a94327aa5d1ba6b46b6ce7 : FAILURE in 31m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a3963a12c6248a7924cd8c23cdde620 : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0eacabd661572cf28f023e88f6c1cb097b9de69","author":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"date":"2020-04-16 19:10:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f4df8dcd06ff793f1882b3e356ab0df8b0291f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 19:52:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98dc89a7750c4dc48c74969e55318e95 : POST_FAILURE in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3876101e63ec49649d94349fb8998695 : POST_FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b949935930407292bb328e3102a7c5 : FAILURE in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/885c2b348acc4369863bb124399ba366 : POST_FAILURE in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74148a72e07846f4b79fdf848e5a4f46 : POST_FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf07523aa4c48c1840f05418a4b5566 : POST_FAILURE in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7517321b63364a8aa016ee7e8dd515f9 : FAILURE in 5m 29s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/9c753a72cc2c4b199d2e6749b5521a65 : FAILURE in 31m 39s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/2ca27a1bec9b405aa1a4d53d3d1618b2 : FAILURE in 33m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbdd9d2ff767435bb469fb0112da8aef : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cffffa1b9e52f0210eeb2848e131222c48340a7b","author":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"date":"2020-04-16 20:18:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d9dd72f6067739c644350988ff31b716abfe7637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 20:54:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/873773a04f164436927f607e3fdc9d5f : POST_FAILURE in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb4bd655317c49a4a4a3645096fa4efc : POST_FAILURE in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8953ce1c4c4cd48decd74367252b9f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1ba809891474aadbb2291cd9a9fd4a4 : POST_FAILURE in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32bedc76a7ef4fba9ce0fabd4fd3e2c6 : POST_FAILURE in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c146155bebd480ba3870fe09a86b43c : POST_FAILURE in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fec29f37244a82b947b87715128033 : FAILURE in 7m 55s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/ff077e3a05b145a58360fda00d2e6321 : FAILURE in 28m 05s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/7c4dca16b99042fdafb8c2a1935a96e3 : FAILURE in 26m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78200f1ba1794ecda51a30c96caa0ac9 : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"32333c53bb619226a01486a63b6a9f98ad6c7707","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 21:10:35.000000000","message":"Patch Set 3:\n\n@Zhuo: Please run unit tests locally first to avoid running the expensive devstack job each time on limited CI resources. See https://docs.openstack.org/blazar/latest/contributor/development.html for instructions.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a580254e42aeb84d5a02fb9b5181a0612445475","revisions":{"1506ffc32d89c76bf5a3a2b143d83a11e80613d7":{"kind":"REWORK","_number":1,"created":"2020-04-16 16:30:46.000000000","uploader":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"ref":"refs/changes/48/720548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/48/720548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/48/720548/1"}}},"commit":{"parents":[{"commit":"738f22c4a9dc2ea30757ee42f12abebe900ea68b","subject":"Fix docs publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/738f22c4a9dc2ea30757ee42f12abebe900ea68b"}]}],"author":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 16:20:10.000000000","tz":-300},"committer":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 16:20:10.000000000","tz":-300},"subject":"Prevent reserving ironic nodes in error states","message":"Prevent reserving ironic nodes in error states\n\nCurrently, Blazar allows reserving ironic nodes in error or maintenance\nstates. This is a problem because nodes in error or maintenance states\ncan not launch instances. Polling-based monitoring is updated to first\nidentify \"ironic\" hypervisor_type. If ironic nodes, pull node states\nfrom ironic and decide \"failed resources\" and \"recovered resources\".\n\nChange-Id: Id3531226dc4d10b46d26820f3f47e2aefb3089a6\nCloses-Bug: #1872741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1506ffc32d89c76bf5a3a2b143d83a11e80613d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1506ffc32d89c76bf5a3a2b143d83a11e80613d7"}]},"branch":"refs/heads/master"},"8ffa352c7de02d26465d05c8a048407e686843ec":{"kind":"REWORK","_number":2,"created":"2020-04-16 19:10:35.000000000","uploader":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"ref":"refs/changes/48/720548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/48/720548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/48/720548/2"}}},"commit":{"parents":[{"commit":"738f22c4a9dc2ea30757ee42f12abebe900ea68b","subject":"Fix docs publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/738f22c4a9dc2ea30757ee42f12abebe900ea68b"}]}],"author":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 16:20:10.000000000","tz":-300},"committer":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 19:10:22.000000000","tz":-300},"subject":"Prevent reserving ironic nodes in error states","message":"Prevent reserving ironic nodes in error states\n\nCurrently, Blazar allows reserving ironic nodes in error or maintenance\nstates. This is a problem because nodes in error or maintenance states\ncan not launch instances. Polling-based monitoring is updated to first\nidentify \"ironic\" hypervisor_type. If ironic nodes, pull node states\nfrom ironic and decide \"failed resources\" and \"recovered resources\".\n\nChange-Id: Id3531226dc4d10b46d26820f3f47e2aefb3089a6\nCloses-Bug: #1872741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8ffa352c7de02d26465d05c8a048407e686843ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8ffa352c7de02d26465d05c8a048407e686843ec"}]},"branch":"refs/heads/master"},"6a580254e42aeb84d5a02fb9b5181a0612445475":{"kind":"REWORK","_number":3,"created":"2020-04-16 20:18:02.000000000","uploader":{"_account_id":29269,"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","username":"zhenz"},"ref":"refs/changes/48/720548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/48/720548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/48/720548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/48/720548/3"}}},"commit":{"parents":[{"commit":"738f22c4a9dc2ea30757ee42f12abebe900ea68b","subject":"Fix docs publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/738f22c4a9dc2ea30757ee42f12abebe900ea68b"}]}],"author":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 16:20:10.000000000","tz":-300},"committer":{"name":"Zhuo Zhen","email":"zhenz@uchicago.edu","date":"2020-04-16 20:17:50.000000000","tz":-300},"subject":"Prevent reserving ironic nodes in error states","message":"Prevent reserving ironic nodes in error states\n\nCurrently, Blazar allows reserving ironic nodes in error or maintenance\nstates. This is a problem because nodes in error or maintenance states\ncan not launch instances. Polling-based monitoring is updated to first\nidentify \"ironic\" hypervisor_type. If ironic nodes, pull node states\nfrom ironic and decide \"failed resources\" and \"recovered resources\".\n\nChange-Id: Id3531226dc4d10b46d26820f3f47e2aefb3089a6\nCloses-Bug: #1872741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6a580254e42aeb84d5a02fb9b5181a0612445475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6a580254e42aeb84d5a02fb9b5181a0612445475"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
