)]}'
{"id":"openstack%2Fblazar~724979","triplet_id":"openstack%2Fblazar~master~I4600211320beeb2b81e620bcce8545cb60b0a3ad","project":"openstack/blazar","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"I4600211320beeb2b81e620bcce8545cb60b0a3ad","subject":"Add py38 package metadata","status":"MERGED","created":"2020-05-02 12:50:12.000000000","updated":"2020-05-11 17:17:28.000000000","submitted":"2020-05-11 17:15:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724979-1589217352797-4f1731f4","meta_rev_id":"c1beff37999f6da23631e61e8bbd925df9da95c5","_number":724979,"virtual_id_number":724979,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 17:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2020-05-09 19:40:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-05-11 16:26:02.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 12:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-05-11 16:26:02.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 12:23:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-09 19:40:15.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-11 16:26:02.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2020-05-11 17:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52bfd7e9103e9409be63b953a81c77a5a0594934","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-02 12:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91e2b5a863741f1441f93e73b13e714005deefd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 13:34:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc363345e09b4f84b815e9384801d21a : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819de80299514b729212a16ed06f0bbf : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc89c26fc44045e6914f890d5287947a : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bf7adee02944c4bdbcdc1e6d46b163 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e605db7b464567ba22f9bf11dd6f53 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6942e7ddcd84fee95dd2b9a9b634b34 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae10ebff13d400c9b969324030baa15 : SUCCESS in 13m 48s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/c59de06bd1af464db5a4d61b71aead18 : SUCCESS in 42m 05s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/dcb3ca18e24f47c9824c19598e87c171 : SUCCESS in 42m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81264fe63ec54fd8b0609d09226d50e6 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc99ac65cdadfa3dd107f5a685abc7e64b8567b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-05 12:23:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb6a2c3c9e50158cd0ca7d817398822849ccdf5c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-09 19:40:15.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":1},{"id":"530e710f97fb92e915b1a3efb4ab67b80a1736ee","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2020-05-11 16:26:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"43598f9b26896729105bd371181deb275dc4dd0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:27:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca02b189048c1bf832f17fe689039b00e8fc925a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:15:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/362835c1c1ab452a80d0c59b5b0c069b : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e8f09f3e67346a5a3c6d1565097730a : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d354eaa3daef4a78abc772e3c6f24498 : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88279ac24c5453eaeaa739f97a822a3 : SUCCESS in 13m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ff76adcb4c4f84bf79338d3fe1e476 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30b1d7a224d46efb998822202e1a7b1 : SUCCESS in 14m 56s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/8ba16e1828f74a19875a471b00fc658a : SUCCESS in 44m 27s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/6034f270f1af4ed980f8f74bf7583e0f : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":1},{"id":"dcead3131ad04fe58938147422e22860a55f9d76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:15:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09452f284242175ac8ad7ae814e9e3d094fb87f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:17:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b7dc907d3442619bd9e85ac2542257 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ff1fb58b35643fe231e811a4b214a876d23bf37","revisions":{"2ff1fb58b35643fe231e811a4b214a876d23bf37":{"kind":"REWORK","_number":1,"created":"2020-05-02 12:50:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/79/724979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/79/724979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/724979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/724979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/724979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/79/724979/1"}}},"commit":{"parents":[{"commit":"aa24d21d173586cde23b446976679bd5cd183f83","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/aa24d21d173586cde23b446976679bd5cd183f83"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 12:48:11.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 12:48:11.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: I4600211320beeb2b81e620bcce8545cb60b0a3ad\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2ff1fb58b35643fe231e811a4b214a876d23bf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2ff1fb58b35643fe231e811a4b214a876d23bf37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
