)]}'
{"id":"openstack%2Fblazar~730706","triplet_id":"openstack%2Fblazar~master~If6f6cf21170e95c58dd8a0a1b41b9b5f4797a627","project":"openstack/blazar","branch":"master","topic":"bug/1880646","hashtags":[],"change_id":"If6f6cf21170e95c58dd8a0a1b41b9b5f4797a627","subject":"Fix up adding availability zone (AZ) to a host on creation","status":"MERGED","created":"2020-05-26 06:31:35.000000000","updated":"2020-07-01 01:11:58.000000000","submitted":"2020-07-01 01:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730706-1593565825423-fca765a5","meta_rev_id":"c6aeec5d481cdd424d9a7812a7acbad783b0869b","_number":730706,"virtual_id_number":730706,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 01:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":2,"date":"2020-07-01 00:29:14.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 09:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":1,"date":"2020-07-01 00:29:14.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 09:42:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-01 00:29:14.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-07-01 01:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c20e77af68f2e087864d0c1d24c9b46b3fa9fc04","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-05-26 06:31:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a88da2244c606e8f12dc85a5875a4082569663bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 07:20:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8002bf0c37264e35beba7584bab91710 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d23cb241b674799b2ae4a00edb65b12 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10cd9eceaa3346e79073bc1c08408fb7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46e083ae930e4fddb89c834b255f5662 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c066b7c799ef45e297bb571d59f74484 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01289c3b81da4f64b994a7273220e78d : SUCCESS in 8m 36s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/b174062179e049c7a37c920f027faac1 : SUCCESS in 42m 46s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/a76e96aa917b47bd8e8f33e5ea9cee67 : SUCCESS in 40m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/739681f30ce0413c936157edb502767f : SUCCESS in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5c7900b6394b77557d449edc62d8be57e5be47e","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-06-23 09:22:42.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"5b1d5dc07a6548c2f2eaeba22cac54237d91c5e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 09:14:29.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"dcbab2c3b2a9e805d73ee0b87c27ccb3435f2d43","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 09:17:56.000000000","message":"Patch Set 1: Code-Review-1\n\nThank you! Let\u0027s add a release note to describe the bug fix.","accounts_in_message":[],"_revision_number":1},{"id":"59b5f782f087c01fad8874404d2d54f4dda9657d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 09:42:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c786f00b4ad5bb8282a63ca34cb7e2cb8bf59ac7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 09:42:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6bf69e7724e74a8db747081d3522bb90b34445ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:38:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03aa15d3c76a4ca28fa05bfd745d431a : SUCCESS in 15m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7adaf789e1bf4da39d9707a9a605a4f5 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b645446c9c3a438b914f1419b83ee0c8 : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742ac6718a45463cbc28a44dae946601 : SUCCESS in 14m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4609d5dcc1f45d9979baf5c556c976f : SUCCESS in 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7adfb02b7440e1b5ec0c0019976767 : SUCCESS in 22m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1b76420b94345ecbd5ec191f6480fc6 : SUCCESS in 9m 22s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/64156cefcaf24f819557ab395a51c4c7 : SUCCESS in 52m 27s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7e903719214cb48571295c0778d100 : SUCCESS in 53m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c30922cdbd4845378c1c83a361d05a4d : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6da5f790c94ef5d994a8d3f44bd351365a3566cd","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-07-01 00:29:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThe release note looks good to me. Thanks, all.","accounts_in_message":[],"_revision_number":2},{"id":"e079f0516bed61a6e4a0eec5bee42c4fb17ee107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 00:29:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfda7700db9f72d452f0a7b696949bbabac384bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 01:10:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94ecebae87234126b5064202872f7d7d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda09c6d3ebe48bc9522c7b20c40851e : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26f529e6612a46f5bc5a66014a03ae6f : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0efc43c314664ac4b822ad7898a68d58 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1088a69d51e420bbe6f75ce2be2514a : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5681f60971c04cfc8b73af365a03f745 : SUCCESS in 5m 37s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/9a4d1a57732f44e09e6311344748b112 : SUCCESS in 38m 21s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/454a0f1298104a989e14474ddb07a71e : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":2},{"id":"7de39c726a818f664b077b20318b727796e32639","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 01:10:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"06d2b48c514b2f19248367f0e9eaf9a77b59ff55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 01:11:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3982e406ed50499eb3d81d2fb4b105d9 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de82f094d3b54ff1a1d2cf3063b6b25f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"532544efbf5fc3a2daf60af5d3c216b0b6a62c42","revisions":{"3784d9613e2f7fd2790bf0c11d813e1311e91559":{"kind":"REWORK","_number":1,"created":"2020-05-26 06:31:35.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/06/730706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/730706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/730706/1"}}},"commit":{"parents":[{"commit":"254218fb6074ede0015d39962b63718f4c82bc07","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/254218fb6074ede0015d39962b63718f4c82bc07"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 06:29:27.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 06:29:30.000000000","tz":600},"subject":"Fix up adding AZ to a host on creation","message":"Fix up adding AZ to a host on creation\n\n1. nova AZ returns hosts as service hosts as opposed to\n   hypervisor hostnames\n2. Handle case where a nova AZ has no hosts\n\nCloses-Bug: #1880646\n\nChange-Id: If6f6cf21170e95c58dd8a0a1b41b9b5f4797a627\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3784d9613e2f7fd2790bf0c11d813e1311e91559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3784d9613e2f7fd2790bf0c11d813e1311e91559"}]},"branch":"refs/heads/master"},"532544efbf5fc3a2daf60af5d3c216b0b6a62c42":{"kind":"REWORK","_number":2,"created":"2020-06-30 09:42:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/730706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/730706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/730706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/730706/2"}}},"commit":{"parents":[{"commit":"9cb7f6b52b9399e748a5c665b1bcae8ee25e9375","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9cb7f6b52b9399e748a5c665b1bcae8ee25e9375"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-05-26 06:29:27.000000000","tz":600},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-30 09:40:43.000000000","tz":120},"subject":"Fix up adding availability zone (AZ) to a host on creation","message":"Fix up adding availability zone (AZ) to a host on creation\n\n1. Nova AZ returns hosts as service hosts as opposed to hypervisor\n   hostnames\n2. Handle case where a Nova AZ has no hosts\n\nChange-Id: If6f6cf21170e95c58dd8a0a1b41b9b5f4797a627\nCloses-Bug: #1880646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/532544efbf5fc3a2daf60af5d3c216b0b6a62c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/532544efbf5fc3a2daf60af5d3c216b0b6a62c42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
