)]}'
{"id":"openstack%2Fblazar~732697","triplet_id":"openstack%2Fblazar~master~I016d0ecf3751212fc578d30e5ba0a5cec48ac252","project":"openstack/blazar","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I016d0ecf3751212fc578d30e5ba0a5cec48ac252","subject":"Use the native Zuul v3 jobs from blazar-tempest-plugin","status":"MERGED","created":"2020-06-02 14:46:32.000000000","updated":"2020-08-18 11:00:07.000000000","submitted":"2020-08-18 10:57:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732697-1597748255960-86100347","meta_rev_id":"7fcfd6bccc08a95fa1e89ea3b3e502b09f7da20f","_number":732697,"virtual_id_number":732697,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"date":"2020-08-18 09:33:33.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 10:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":2,"date":"2020-08-18 09:07:48.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 17:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":1,"date":"2020-08-18 09:07:48.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-14 08:11:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 08:11:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-18 09:07:48.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-08-18 10:57:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"549a9da7aebb8717ba6e26b71966aaae333013e1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-02 14:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359e8ac60542e3403a5aba34772e4c711f1900e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 15:31:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a47ceb11f50144b08c4067d70dc14d49 : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b66e74c6d354760a535ea9e5d9d5e96 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31304f872ca240c5a0b6f5f70ebc9c9d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d00e6ec01f8482c8eb789939f866c92 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/365958c8d7064c1faa00b9ae957b8e13 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8801b291fb4356b10d414830b5d16a : SUCCESS in 6m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/fef16b8d360a4a4e924ad15119bc0ba2 : FAILURE in 41m 04s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ad51d3c73a0b40ba89aa154172d81bb9 : FAILURE in 39m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b374afa3522448e0ac28f452e76a4b12 : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c44f5e891f4b5d171a0c3ec1a6e80d878448c978","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-24 08:49:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de102bcd7d7aae7ba9fb19bae26848f9ab9f3fb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 09:33:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21318b172f0d4fb0b6620f6146171d7d : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b406ae9415cc48c3ab9276b140dd9187 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb57fc27b104fd7a930cdab9f53fd4d : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81ac3dda96d74a07a55d2f78904838ca : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129e5a8a61b14eef951220660e87d6af : FAILURE in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b6b4e0eee24f199709e26c821b689b : SUCCESS in 9m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/987787969ccb49c7a07f381319318504 : SUCCESS in 40m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/2877ffbfa2c64347a1cbb7fe75a05fb2 : SUCCESS in 40m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac33103d1a5041fbaec4afd490785abf : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b8299a3bd7aeafbea1b8d5b152dada499cffc4d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-24 09:48:34.000000000","message":"Patch Set 1:\n\nIt looks like the py38 failures is unrelated to this change.","accounts_in_message":[],"_revision_number":1},{"id":"60d99539cf4ae6bd8206736802bcc5d53d639291","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-13 16:43:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d2868aaa8b72d083f72b5c0d99de10f7a21c86d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 17:38:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/672a1cdca2364d918f78fd7130c87db1 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36267d91b8c42f18d7112bd980d6853 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0966bc6d96bb4bc4b317c6d72130cb5a : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a431c13b9bf14f09a2695f11e7f2d8ba : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18086aba7a544ae88c1827af9f41fa1b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a256b26bbc734200a81190e1b6869e71 : SUCCESS in 7m 37s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1f117fd441124b768b3359136895f9d5 : SUCCESS in 42m 31s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8e1b29c1979149f0a76767b31ef72705 : SUCCESS in 46m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7434ce935aec4c60affea2f0487ef4f7 : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8dd09874c88eea19aa631d3bfc07cf2e07509a5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-13 17:45:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"210bfe69e66d2e31170cc5a01db8a0dc604c2200","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-14 08:11:07.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"c9ef17f6b054dd8d79c8d678213a33983f8d6ef4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-08-18 09:07:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3c2787740e7b7e42efc235e7827bfd718d3cea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 09:08:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ae863ef76e278991655965aac7aa2b8085006f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 09:30:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cabbc3fb8d243f58d8520b4f1addba6 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c22414e86884f509aaf65daf002033f : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d20c5a271d244844a6d9df5d237e8b5f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/813161533eb0461093c4eb7e214d48f6 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0a54145e8f41fa97274e0920e1a19a : SUCCESS in 7m 13s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7ce8bd56d0204d4da4489f0d18e29cb3 : FAILURE in 14m 15s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1260447b02a0400eb1d89ca5fad7f6a5 : FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":2},{"id":"37954f15d85159f137cc7a158a3e9a216d226105","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-18 09:33:33.000000000","message":"Patch Set 2:\n\nrecheck\n\ntemporary glitch due to a new glance_store release not found on some pypi mirror","accounts_in_message":[],"_revision_number":2},{"id":"596bec3ec1af9c03b4c3f508f767ac66d2cd5af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:05:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b4e71c472dd405f9f48133d5307df1d : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00bb43b6c3a24482ba77d40849d02263 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e5c6103c64491787ef16fc5d4dc81c : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aab6f2a030a41799b670a088b52ff18 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467c35d054a14005afbe31587188449e : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66504e820b2b41399189d6cf3d3b0236 : SUCCESS in 5m 37s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ec17eb7a31d34a51b57a2cf4a632777b : SUCCESS in 29m 29s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ca677800542140f1bb02fcdb76c116d5 : SUCCESS in 30m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92ba6193bbde44a886348a6dad33f485 : FAILURE in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fda36adfa9b3c5e82c11d4a7edd1ea81582fbe84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:06:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25684b950b40fe831dc9a07610f33e07e956ab53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:57:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b2ee1e3786e5d0ec4cb9166f4ada42b8c406519b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:57:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ba06f4bac949ddb5f8af2ca1c1068a : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c976e2afafd469aa2736f25fc9d603c : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff2edb155404c448d4dbb11f606641f : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8587819112a046628f659260008ab4d9 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1707abf0a9af4fdea99704d3733f2ee3 : SUCCESS in 9m 19s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4bf805117a5e413b9c1987a931f831fc : SUCCESS in 49m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8f123a9bcf4389b3487852c231bbb8 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b070f2748dcd664490e701733ef50e5f9ee56d96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:00:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57bd3342332442c82fd7b08f1830796 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2f792629c2a6ef70ec6fa6efaa05b2357bff96a","revisions":{"0df5e8280f66b2487b48245e854e60456e22546d":{"kind":"REWORK","_number":1,"created":"2020-06-02 14:46:32.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/97/732697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/97/732697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/97/732697/1"}}},"commit":{"parents":[{"commit":"254218fb6074ede0015d39962b63718f4c82bc07","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/254218fb6074ede0015d39962b63718f4c82bc07"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-06-02 14:42:03.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-06-02 14:42:03.000000000","tz":120},"subject":"Use the native Zuul v3 jobs from blazar-tempest-plugin","message":"Use the native Zuul v3 jobs from blazar-tempest-plugin\n\nThe new base jobs moved to that repository.\n\nDepends-On: https://review.opendev.org/732694\nChange-Id: I016d0ecf3751212fc578d30e5ba0a5cec48ac252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0df5e8280f66b2487b48245e854e60456e22546d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0df5e8280f66b2487b48245e854e60456e22546d"}]},"branch":"refs/heads/master"},"a2f792629c2a6ef70ec6fa6efaa05b2357bff96a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-13 16:43:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/732697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/97/732697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/732697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/97/732697/2"}}},"commit":{"parents":[{"commit":"690ab0972f22372a17ae2a0c700cb4f7195a838f","subject":"Use flask\u0027s builtin methods for test context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/690ab0972f22372a17ae2a0c700cb4f7195a838f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-06-02 14:42:03.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-13 16:43:29.000000000","tz":0},"subject":"Use the native Zuul v3 jobs from blazar-tempest-plugin","message":"Use the native Zuul v3 jobs from blazar-tempest-plugin\n\nThe new base jobs moved to that repository.\n\nDepends-On: https://review.opendev.org/732694\nChange-Id: I016d0ecf3751212fc578d30e5ba0a5cec48ac252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a2f792629c2a6ef70ec6fa6efaa05b2357bff96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a2f792629c2a6ef70ec6fa6efaa05b2357bff96a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
