)]}'
{"id":"openstack%2Fblazar~736450","triplet_id":"openstack%2Fblazar~master~I3c31c744f250ae9d96bc828709be99c9a1efce8b","project":"openstack/blazar","branch":"master","topic":"optimization/pip-no-cache-dir","hashtags":[],"change_id":"I3c31c744f250ae9d96bc828709be99c9a1efce8b","subject":"optimize size and time using \"--no-cache-dir\"","status":"ABANDONED","created":"2020-06-18 06:27:38.000000000","updated":"2020-06-18 12:24:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d369c3f53c68bc93fe6d8d853508a6240108d9f","_number":736450,"virtual_id_number":736450,"owner":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-18 06:57:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 06:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96d497b0030ff61955ecb7a8f0134b8490ed28fe","author":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"date":"2020-06-18 06:27:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cf7c3e73abcf54538bf208ab2315dc6b8ae1db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:57:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a710fb84ea294aaeb633e241095e27d3 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d77a2c288746fab79df6411e8754d3 : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f336e089be534b1fb9b3f9eeaf77f864 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56457427b4c24137aec9ac26048de010 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbe9bc7f0464c24a19da875abbcb03d : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a52927aaa29413fa530dcc26047ef7f : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a42c983e98b4b3690510515e3b6dcce : FAILURE in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a9dc0e9174446108d811d3463caff31 : SUCCESS in 6m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81fd966173354e5799199c564af7a61d : SUCCESS in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d2e28582197cdf3ed6bbf09f1b9e8a3d39a897c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 12:24:22.000000000","message":"Abandoned\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015515.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f40ba2ad2ba8c7efbce347de0df018f574cd29ea","revisions":{"f40ba2ad2ba8c7efbce347de0df018f574cd29ea":{"kind":"REWORK","_number":1,"created":"2020-06-18 06:27:38.000000000","uploader":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"ref":"refs/changes/50/736450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/50/736450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/736450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/736450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/736450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/50/736450/1"}}},"commit":{"parents":[{"commit":"0c3ac1911dcbcb1291001883af0d3ff42654c77a","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0c3ac1911dcbcb1291001883af0d3ff42654c77a"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 04:39:07.000000000","tz":330},"committer":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 06:27:31.000000000","tz":330},"subject":"optimize size and time using \"--no-cache-dir\"","message":"optimize size and time using \"--no-cache-dir\"\n\nUsing \"--no-cache-dir\" flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: I3c31c744f250ae9d96bc828709be99c9a1efce8b\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f40ba2ad2ba8c7efbce347de0df018f574cd29ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f40ba2ad2ba8c7efbce347de0df018f574cd29ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
