)]}'
{"id":"openstack%2Fblazar~738071","triplet_id":"openstack%2Fblazar~master~I93a945f10cf2b8d8c16d43ee80f9a03d8acf4328","project":"openstack/blazar","branch":"master","topic":"bug/1883468","hashtags":[],"change_id":"I93a945f10cf2b8d8c16d43ee80f9a03d8acf4328","subject":"Fix uwsgi path location","status":"MERGED","created":"2020-06-25 17:22:51.000000000","updated":"2020-06-26 12:10:16.000000000","submitted":"2020-06-26 12:06:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738071-1593173174609-b1799fbd","meta_rev_id":"567727b33cbd311eab2fa0281690cee50870c180","_number":738071,"virtual_id_number":738071,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 12:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-06-26 11:16:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-06-26 11:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 11:16:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-26 12:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99296d58eca21e7116a793a63f26da7aa01ad692","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 17:22:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4d1842a82dd32e4e2100c62ab070596ee76f4ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:11:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24042aa1c1764596bdcadae40be44608 : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbec9f6934d04465831b4e9344a793f8 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c39c3d6e3c4c25b9a2c2a4d792f903 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3294dd57a71447d3a207b2ac0b4974b3 : SUCCESS in 11m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a2b15094a7a464981598206b576be3a : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fec02836f8c431588852b16e47fcb2c : FAILURE in 7m 17s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/197cc11b9f2c4b2c91b6157a419af63f : SUCCESS in 42m 55s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/1e2d6ae2a65848fa90c5bb1e1c97e762 : SUCCESS in 40m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ac2e8b7cfb449598a7487a0cb630b6f : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12328d9946ca24b6624136ab6a26e6e85175e612","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 19:09:52.000000000","message":"Patch Set 1:\n\nintegration jobs are good now. doc is failing for pdf build.","accounts_in_message":[],"_revision_number":1},{"id":"455e08a4b4ca4024814f35b917057ee3d53677a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-26 10:14:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83dded0c35906da19cccdd0316228938851b20a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 11:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eca9d786a584577a3e47106edc4c386 : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b906da2cd634f34bfe1b3ca13ec258b : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06a033becd145e2894ad90017003918 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8499ec0a4907446c98e27d2da64a2905 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/927d0566d335488fb8fb54aaacd23323 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc1f51d63c84a87bdaf4142525dffb1 : SUCCESS in 8m 07s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/33d0c6e47ef2407083ea23fd0c6a7b44 : SUCCESS in 44m 01s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/bb82b8de065140b79c562c065f6aa346 : SUCCESS in 42m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e23f41297ab54dfd8d70789548b2f2a6 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb4a90e5279b531ded7ec35a519736e8e356a47b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-26 11:16:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks gmann. Approving to fix the gate.","accounts_in_message":[],"_revision_number":2},{"id":"2d65315280a671059f1bc2ee858fc0be7d334615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 11:16:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb0de5337d0db273fa55a2691b36842bed769cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:06:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c904dd44d69e40d9bcbe43aa46e929cc : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33dad6118a64bf395c79331daeb915c : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3e32d56efb475f9889b20ba4fa9cab : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c225ff6443d4f9f861a36b49181cce9 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b4752d6921447b831253a3f40f4cc3 : SUCCESS in 6m 58s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/640143a06f324cbfba1d901c8cad19b7 : SUCCESS in 40m 46s\n- blazar-devstack-dsvm-ipv6-only https://zuul.opendev.org/t/openstack/build/c7a1c4c5344b490dbea721d002c1c3f1 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":2},{"id":"714642f66a1694a4f7090e3a03f8491c84ead6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:06:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b9c4bbef6dc32a2dca28e07d4a3d146ebafaafe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:10:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3205001856d04c73a0288ec8ce06e7d0 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca1e5ffc623dd57123969883e0311b446b577108","revisions":{"96f5097e46ca9d8019e2d8551c2d69878bf21f28":{"kind":"REWORK","_number":1,"created":"2020-06-25 17:22:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/738071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/738071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/738071/1"}}},"commit":{"parents":[{"commit":"0c3ac1911dcbcb1291001883af0d3ff42654c77a","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0c3ac1911dcbcb1291001883af0d3ff42654c77a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:19:07.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:19:07.000000000","tz":-300},"subject":"Fix uwsgi path location","message":"Fix uwsgi path location\n\nAfter fixing the uwsgi installation from source and uwsgi path\nlocation on devstack side, we need to apply the same fix in\ndevstack plugins side too.\n\n- https://review.opendev.org/#/c/577779/\n\nChange-Id: I93a945f10cf2b8d8c16d43ee80f9a03d8acf4328\nRelated-Bug: #1883468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/96f5097e46ca9d8019e2d8551c2d69878bf21f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/96f5097e46ca9d8019e2d8551c2d69878bf21f28"}]},"branch":"refs/heads/master"},"ca1e5ffc623dd57123969883e0311b446b577108":{"kind":"REWORK","_number":2,"created":"2020-06-26 10:14:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/738071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/738071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/738071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/738071/2"}}},"commit":{"parents":[{"commit":"fe0fea186a1d42696ed8c7e83755bb1a991addee","subject":"Fix pdf-docs build failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fe0fea186a1d42696ed8c7e83755bb1a991addee"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:19:07.000000000","tz":-300},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-26 10:12:41.000000000","tz":120},"subject":"Fix uwsgi path location","message":"Fix uwsgi path location\n\nAfter fixing the uwsgi installation from source and uwsgi path\nlocation on devstack side, we need to apply the same fix in\ndevstack plugins side too.\n\n- https://review.opendev.org/#/c/577779/\n\nNOTE(priteau): Enable Zuul jobs that were disabled in commit\nfe0fea186a1d42696ed8c7e83755bb1a991addee now that they are passing.\n\nChange-Id: I93a945f10cf2b8d8c16d43ee80f9a03d8acf4328\nRelated-Bug: #1883468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ca1e5ffc623dd57123969883e0311b446b577108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ca1e5ffc623dd57123969883e0311b446b577108"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
