)]}'
{"id":"openstack%2Fblazar~753263","triplet_id":"openstack%2Fblazar~master~Ia13402213bc82027e3d658918a22003be26576d6","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"Ia13402213bc82027e3d658918a22003be26576d6","subject":"Make test-setup.sh compatible with mysql8","status":"MERGED","created":"2020-09-22 09:10:33.000000000","updated":"2020-09-22 12:05:49.000000000","submitted":"2020-09-22 12:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753263-1600776252680-72df3f54","meta_rev_id":"8153d7d19d4301e95a7c8b1f1c051ade24835149","_number":753263,"virtual_id_number":753263,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 12:04:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":2,"date":"2020-09-22 11:15:56.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":1,"date":"2020-09-22 11:15:56.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"all":[{"value":2,"date":"2020-09-22 11:15:56.000000000","_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-22 09:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 11:15:56.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2020-09-22 12:04:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9988df4478de566a1f830a417c7a0fc0cc1e6f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-22 09:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e09fd5037b960c50eeea237a9504f0ea0a5a2801","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-22 09:19:47.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c530791aa1df3e87c9104db2bc88d6c14a6a6aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50c6e122290f4a85a29ff3801b16a769 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68c7c130c0a64e099859a11021d97b3e : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f8c74602cf4fbebbe3a58f1088eb55 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f12c2a468213465a9a5529c6b6850991 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b3fa31267f646908c91362ba3b0aa58 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85dec84e907d4d88892d6ef337fe59c6 : SUCCESS in 20m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/595ea7ab5574459ba5f63e58c65d7d43 : SUCCESS in 44m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/2aa11be9bff6466f9e31ceedcd89d4c9 : SUCCESS in 41m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bbccbe28994e46cf85b47a573892c38d : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecbddd992fa67aa3184c957ef14cb06e74261174","author":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"date":"2020-09-22 11:15:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"bba9e1f2637f012a891a893939ec733d88d1db6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:16:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd649195cae702bd53351e5e503dfc1b063da6a5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-22 11:46:59.000000000","message":"Patch Set 1:\n\nThank you Bertrand :-)","accounts_in_message":[],"_revision_number":1},{"id":"e29dbe51ec48b1b2d699e6aca840baa2327da324","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:04:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b44c9d3a99e944068706c44afbbbdb0e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43c2a5a18cc43d2bb8e60a483654542 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f704098cc59044b7a2a43ddfb104273a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae6655b22b04593848882e9a0f79452 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0810cf595394490bf741d2f94f23463 : SUCCESS in 19m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/416524409d4349ef8625c492ad075213 : SUCCESS in 41m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ebbd95aff33b4146a7b133ab7c245f15 : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"145bff1f951706e4c74b1ece41eff862b1909607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:04:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8153d7d19d4301e95a7c8b1f1c051ade24835149","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b23c3835f14a85b381c062019908a7 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae22b23906a93a83a95154de8ee57381b61f3597","revisions":{"ae22b23906a93a83a95154de8ee57381b61f3597":{"kind":"REWORK","_number":1,"created":"2020-09-22 09:10:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/753263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/753263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/753263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/753263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/753263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/753263/1"}}},"commit":{"parents":[{"commit":"a2f792629c2a6ef70ec6fa6efaa05b2357bff96a","subject":"Use the native Zuul v3 jobs from blazar-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a2f792629c2a6ef70ec6fa6efaa05b2357bff96a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-22 09:09:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-22 09:09:15.000000000","tz":120},"subject":"Make test-setup.sh compatible with mysql8","message":"Make test-setup.sh compatible with mysql8\n\nStarting from mysql version 8 it\u0027s not possible to create a user\nimplictly when using GRANT.\n\nThis patch makes the behavior compatible with that.\n\nChange-Id: Ia13402213bc82027e3d658918a22003be26576d6\nCo-Authored-By: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae22b23906a93a83a95154de8ee57381b61f3597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae22b23906a93a83a95154de8ee57381b61f3597"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
