)]}'
{"id":"openstack%2Fblazar~755166","triplet_id":"openstack%2Fblazar~master~Id572aec8a6fda7676eb03d17c66fd33f9801f25d","project":"openstack/blazar","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"Id572aec8a6fda7676eb03d17c66fd33f9801f25d","subject":"Remove six","status":"MERGED","created":"2020-09-30 03:00:55.000000000","updated":"2021-01-05 18:59:09.000000000","submitted":"2021-01-05 18:57:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755166-bp/six-removal","meta_rev_id":"fa8dfa72f683834fc1e52e3b5dce4ecfd9b2bf2a","_number":755166,"virtual_id_number":755166,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 18:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":2,"date":"2021-01-05 17:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":1,"date":"2021-01-05 17:55:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 07:24:03.000000000","updated_by":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"reviewer":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"state":"REVIEWER"},{"updated":"2020-10-02 10:22:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-07 03:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 08:31:07.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-12-22 03:38:12.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"0a2363b20e9fd8c916089cf649af5f384b9042fa","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 03:00:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80eabc6a2469db7f494ed6a1e9eba3b23bb59300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 03:49:59.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4b84f6ea0e74615b43e2ca4bfe212bd : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62222f22e02a4b4b9bd90a37e71c7dfd : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a35ca93dc54aa698c8dc98ced84739 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b3c90a735b44e5affbbbfd71979485 : FAILURE in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89d02614e7a544b8994ce852de1307ce : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111e3771e82a41629b5ed0713c4bea2c : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4ea2b670ac4f12b681384d812c7305 : SUCCESS in 8m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/744f9784251e4b92a7570c551120baac : SUCCESS in 37m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/d77b2ef08df04046ab1d9e4eb6fc0642 : SUCCESS in 47m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e71e1e527c5417e927bcae2ccd6d677 : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd79949f5ca4d9287aac0e02f1e7076c36b8dea7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 06:00:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"779c5e19d76aabeb598b6e7fbfe25d262d3c401c","author":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"date":"2020-09-30 06:21:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0a8893ec95f3f1e5eb2318784d9cc4b12a7c96a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 07:22:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3635432f655742c8914d8abdc54b42b3 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc50b11d5c41480fa92d91a7b92cde46 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1177d6e384924b8cb5ef918a97ca2ef3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8896e52dc9024b60ade44d1bc6222c15 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbf83551a45940ac8c244382de782be9 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7396298577904d8eb934837919731b98 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c06dec1185842be85253a9aa4d2130b : SUCCESS in 7m 05s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d44063190498446a92af170d5c615eb4 : SUCCESS in 40m 25s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/36319be00118416a9246149ca668eaae : SUCCESS in 34m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b36b893cea14bd6a01c076499517c44 : FAILURE in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2a685ee8c1110604f47de832486974d01b3349f","author":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"date":"2020-09-30 07:24:03.000000000","message":"Patch Set 3: Code-Review+1\n\nLSTM","accounts_in_message":[],"_revision_number":3},{"id":"e13a6e724145418ea88ccf9a9dfff63a4026ab8b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 08:22:43.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4c89bd935b386468c49ea12e133cca3325323a27","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 08:26:28.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":4},{"id":"d1f454c7c3a6466413578dcbf9205f46ec1192e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:16:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07152fd34f4544cfa438f7fef9f790d0 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/094efa1edfea49c589acb284914ddd11 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd152dab47ed4430a2f04571f45bb619 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680a0a3e8adc42419f757c98f2121a9e : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/425bb296e6854b1183e414ac5bac662d : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3170c167a48f4ed9b95e791b13d17252 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3078956abd34aeaab27a2cedca527fe : SUCCESS in 8m 23s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cea033a2e90144ddb62ac287f6547d01 : SUCCESS in 43m 03s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4f60e6116d5c46989cbf6c12cc717212 : SUCCESS in 46m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/561307179701465d8108a8b7a8d831e0 : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c505c33f405e20a0fa23a2c62113747616a8df32","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 10:22:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nThanks. I left some comments inline, could you please update your patch?","accounts_in_message":[],"_revision_number":4},{"id":"53ab307073eaaae0b4e3d9b3b7b1feffb5cc44b5","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-07 03:03:56.000000000","message":"Patch Set 4:\n\n\u003e (4 comments)\n \u003e \n \u003e Thanks. I left some comments inline, could you please update your\n \u003e patch?\n\nOK， thanks for review","accounts_in_message":[],"_revision_number":4},{"id":"a6f68e65810867ea29356817a59d4174d97028a2","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-07 03:04:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9675a5101e397559d646147aae0a0e6b65ba0dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 03:50:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cddb06eba2214921ba3ece3a780f427c : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c91adba467342dd9d4c320c50bc6381 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3426dd99c9fd4d7d87771a92dd58320a : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549d8695242c4065a902d9a3636cc477 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b6590b2d0344eeda4494af02240e323 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcb99031d860487194b1afcfc3a4ddce : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc9e861f6b442a39c5cc6d02b437145 : SUCCESS in 7m 55s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1e33e1c07ccd4cd3a6783cca2273c8da : SUCCESS in 42m 00s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e97fcffaa49d4b3a93d80ab9c4115827 : SUCCESS in 41m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6da27b68ac5e4d09b52c2e202b567c5d : FAILURE in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72274731a5bb8ffb44dae425aac31e528db1ba7b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-09 08:31:07.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"a66a8c7d0e7d16bc4d5ec56ae408715f00d6ea1a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-12-22 03:38:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"56185be3344ea0585a9fa72c51fcce9fc7b4c1ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 16:07:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bfe9c2ea90d70ecaf5fcfe51fb0b148f673bcce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 16:08:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"88a30499e65d495dad146aba22cd2f90688b8c10","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 16:08:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"29829afbcc3d3ed743e675a7d7be169e31e720bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 17:33:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d519d13c766e4e74886a0c5f89d2d464 : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9c3bddd56fb4e60b2b04f6c26b2e609 : SUCCESS in 16m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c42498c8aeb41dbb40ccbd062addf0f : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f53da0e9e046e78cb3d7b895af51cd : SUCCESS in 16m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd1c4073dcc42c189c24cdcde692659 : SUCCESS in 16m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9095fb8e689b453898a61dd0698ef6a8 : SUCCESS in 16m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b70c6e6e0dbc4eacb52cdb3b9b9bfc00 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b512a73cb5994678935996a0dbca9468 : SUCCESS in 26m 15s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7f6e98ca55774b49802da16e2165b35c : SUCCESS in 1h 00m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ba389a490f6c40709b8473eb993b2edf : SUCCESS in 56m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f4fc4f5811e431486f0b5149b3d4246 : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e06c68c47aefc27d2deca7fcd75cb11afe28cb0f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 17:55:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"80880ff72778f4bf18acbbfbeb76ded73f9d8668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 17:56:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"33ba7ecfab17ebfbcc5a186c4b2231bf4ecd3587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 18:57:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b62d83f822b94aef8aee2cd3a9980e40 : SUCCESS in 2m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/298621f936b0468d8912995d49d6c695 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130a60f668484601818f8563760fc2b9 : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f5fd41c3b447c28fe019cc7fc2c058 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c80945e3652f492e97fd44a6983d5973 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5de9887afb4e3491035897dc835083 : SUCCESS in 7m 14s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4e1e76be6e46461796ce4202fc2c5978 : SUCCESS in 39m 01s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/874cf22e84b34fdaa14eb8364cc8b830 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":6},{"id":"4d4595f7f5cbb785bbb2ae2e994e4e234662839d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 18:57:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"fa8dfa72f683834fc1e52e3b5dce4ecfd9b2bf2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 18:59:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63fdb46c0d64bf89d942fb6c1639e83 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ae43ea3adf2651d0ba77066053d4e73fa1507aa8","revisions":{"112d84e1597fbe529976922d09b4959ab86b8cbf":{"kind":"REWORK","_number":1,"created":"2020-09-30 03:00:55.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/755166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/1"}}},"commit":{"parents":[{"commit":"021d1d7a17eedc9b4309e3d5ae7b0a42594e952f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/021d1d7a17eedc9b4309e3d5ae7b0a42594e952f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"subject":"Remove six Replace the following items with Python 3 style code.","message":"Remove six\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/112d84e1597fbe529976922d09b4959ab86b8cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/112d84e1597fbe529976922d09b4959ab86b8cbf"}]},"branch":"refs/heads/master"},"489d7a1f8d95edf84552b0aa90d57d3c1733f7c0":{"kind":"REWORK","_number":2,"created":"2020-09-30 06:00:45.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/755166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/2"}}},"commit":{"parents":[{"commit":"021d1d7a17eedc9b4309e3d5ae7b0a42594e952f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/021d1d7a17eedc9b4309e3d5ae7b0a42594e952f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 06:00:36.000000000","tz":480},"subject":"Remove six Replace the following items with Python 3 style code.","message":"Remove six\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/489d7a1f8d95edf84552b0aa90d57d3c1733f7c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/489d7a1f8d95edf84552b0aa90d57d3c1733f7c0"}]},"branch":"refs/heads/master"},"ad6d7a7a93bc6653418273fb5ef5d4674f064618":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-30 06:21:10.000000000","uploader":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"ref":"refs/changes/66/755166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/3"}}},"commit":{"parents":[{"commit":"021d1d7a17eedc9b4309e3d5ae7b0a42594e952f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/021d1d7a17eedc9b4309e3d5ae7b0a42594e952f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"Qitao","email":"15662728521@163.com","date":"2020-09-30 06:21:06.000000000","tz":0},"subject":"Remove six","message":"Remove six\n\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ad6d7a7a93bc6653418273fb5ef5d4674f064618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ad6d7a7a93bc6653418273fb5ef5d4674f064618"}]},"branch":"refs/heads/master"},"06fafeedbb990fe5f585780f8663128ac26bc050":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-30 08:22:43.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/755166/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/4"}}},"commit":{"parents":[{"commit":"021d1d7a17eedc9b4309e3d5ae7b0a42594e952f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/021d1d7a17eedc9b4309e3d5ae7b0a42594e952f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 08:21:59.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nImplements: blueprint six-removal\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/06fafeedbb990fe5f585780f8663128ac26bc050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/06fafeedbb990fe5f585780f8663128ac26bc050"}]},"branch":"refs/heads/master"},"a4e26c627889b2f962a0848db4d19b55f46d7660":{"kind":"REWORK","_number":5,"created":"2020-10-07 03:04:43.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/755166/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/5"}}},"commit":{"parents":[{"commit":"021d1d7a17eedc9b4309e3d5ae7b0a42594e952f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/021d1d7a17eedc9b4309e3d5ae7b0a42594e952f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-10-07 03:04:26.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nImplements: blueprint six-removal\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a4e26c627889b2f962a0848db4d19b55f46d7660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a4e26c627889b2f962a0848db4d19b55f46d7660"}]},"branch":"refs/heads/master"},"ae43ea3adf2651d0ba77066053d4e73fa1507aa8":{"kind":"REWORK","_number":6,"created":"2021-01-05 16:08:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/755166/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/66/755166/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/66/755166/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/66/755166/6"}}},"commit":{"parents":[{"commit":"4f642c877f46646b22602ea495d9e265c7887c62","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4f642c877f46646b22602ea495d9e265c7887c62"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 03:00:38.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-05 16:08:14.000000000","tz":60},"subject":"Remove six","message":"Remove six\n\nReplace the following items with Python 3 style code.\n\n- six.next\n- six.text_type\n- six.add_metaclass\n- six.moves\n\nImplements: blueprint six-removal\n\nChange-Id: Id572aec8a6fda7676eb03d17c66fd33f9801f25d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae43ea3adf2651d0ba77066053d4e73fa1507aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae43ea3adf2651d0ba77066053d4e73fa1507aa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
