)]}'
{"id":"openstack%2Fblazar~755174","triplet_id":"openstack%2Fblazar~master~I6598a8e91536bf17b1877ee136f2df3118f5cebb","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"I6598a8e91536bf17b1877ee136f2df3118f5cebb","subject":"Remove install unnecessary packages","status":"ABANDONED","created":"2020-09-30 03:45:35.000000000","updated":"2020-10-12 08:02:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"174a9c446376dd8468028c73de6bc59134207698","_number":755174,"virtual_id_number":755174,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-30 04:30:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-10-02 10:25:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 04:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 10:25:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"40f53b9932217708338bef9cbc7054cf707192a0","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 03:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44f8b4b55298a3875fe9e99a863c5a242d19995b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 04:30:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b97db79be71744889039c57c858e03f6 : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1afd14feffbe40a1b60aa9a464f65274 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbf33fca61d54e27921b7d173c047d88 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b7ac50bd854a34b0e4ab0f59c36ac6 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef9a975124e948b185f3c78343c24a5c : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c408982872b34588b05921152720a269 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f37973d43374592a912eddbc74c7264 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32f5c9b09b2f48ecaa8457182ea21b8d : SUCCESS in 4m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/9a58da1d610940d094972eb49d9a3f41 : SUCCESS in 40m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9be7923715ab4ff79d21cad52bc5e346 : SUCCESS in 43m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/643133bb5a934787a9adf74220a77848 : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3497a0cf28a55850defce04ddd3a505a0f34bcda","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 10:25:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1184d7d77e87460e3f12286b81cc8e9214cea7d3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-12 07:36:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"174a9c446376dd8468028c73de6bc59134207698","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-12 08:02:36.000000000","message":"Patch Set 1:\n\n@wu.shiming: My -1 vote is a request to update the patch. I would still like to merge it, I was just asking for a minor change to the commit message.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ebf0c26ca3df8754ca69f378ba756a5b445a9be","revisions":{"8ebf0c26ca3df8754ca69f378ba756a5b445a9be":{"kind":"REWORK","_number":1,"created":"2020-09-30 03:45:35.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/74/755174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/74/755174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/74/755174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/74/755174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/74/755174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/74/755174/1"}}},"commit":{"parents":[{"commit":"db8a3752b0847b67f46e2927a6ebe48353e08389","subject":"Remove Babel requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/db8a3752b0847b67f46e2927a6ebe48353e08389"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 03:44:20.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 03:44:20.000000000","tz":480},"subject":"Remove install unnecessary packages","message":"Remove install unnecessary packages\n\nThe docs and releasenotes requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: I6598a8e91536bf17b1877ee136f2df3118f5cebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8ebf0c26ca3df8754ca69f378ba756a5b445a9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8ebf0c26ca3df8754ca69f378ba756a5b445a9be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
