)]}'
{"id":"openstack%2Fblazar~757372","triplet_id":"openstack%2Fblazar~master~Idc9b2ef918cf15472189c4c889752d98d9857be2","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"Idc9b2ef918cf15472189c4c889752d98d9857be2","subject":"bump py37 to py38 in tox.ini","status":"MERGED","created":"2020-10-12 02:54:18.000000000","updated":"2020-10-12 10:21:01.000000000","submitted":"2020-10-12 10:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757372-1602497962055-ed338f59","meta_rev_id":"398889ada87e15a046f30188008523b8189a8bc4","_number":757372,"virtual_id_number":757372,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-12 08:26:50.000000000","_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 10:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-12 08:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-12 08:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 08:43:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-12 10:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"311282afd5b766e3f17f0ec109aa7858af865429","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-10-12 02:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69873260a4abb2c05e50470c02ad796e371f243c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 03:46:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/462294c8152b4d638199aa40437d3cf5 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32fcc0ad2e9c45d3b45837012571b68d : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bda866560c8e41159b29b979cf024cd2 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b10bf52f0c4a27a273dd77b21aa526 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f9a104fdc44b05b34de6e67cd3ef0f : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e0e84250fd4de89f1825ec5963acc3 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74748c0d461497bac265ede9990310f : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35d78d26081d4f6eb40c56d61e376ab5 : SUCCESS in 4m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/afc025e794bd457bbe60300f8b849338 : SUCCESS in 36m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a9736ba0df9742309f037a174d53a515 : SUCCESS in 46m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/994409f7814c4cc4a814ab6b8e54ec00 : FAILURE in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45c89fd6ddd42f6cc286edcc3182a583ec88e6a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-12 08:10:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d79e38a4f30b0616a09bc34dafea44976bace1ef","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-10-12 08:23:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbd873abd30684d94769533b6d9f0f1a36513331","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-10-12 08:26:50.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for your review, Pierre Riteau. Yes,I keep the py36.","accounts_in_message":[],"_revision_number":2},{"id":"c282a4e060bf31aedeaad6ec183148c9c8759f35","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-12 08:43:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3feaeb6a094e42452723ff76e47fb7adc8ea280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:26:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f35396cb0a84d4ba2997b884b8d3009 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4419f5edd1dd4b1eb5ca6ee95719ba05 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2aa55f122074fa1a497cc33b1500621 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07231038c9a440148d92f944334ea663 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2865f08b1e3d458bb7dce24a88124eb6 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/538a90cb5ee0484f8d92f534a387fdb9 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51526eb39edd428b8d2167cff72e2bdc : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbd1a96be2074b9495be57438ef47d8b : SUCCESS in 4m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/dc652dc31653460d83c9380ea345f176 : SUCCESS in 47m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/321bfbb1134742b595e0ab7cfd5f7440 : SUCCESS in 47m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fcef1e6858a42f0ad2ff0e917e91ba1 : FAILURE in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1a6f0e177e5388b406fbbe6d547c056da103975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:26:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"319b3a8df780a9407cf14b91312ad291344f173f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:19:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e85623f96a540edb47eaf1d422630a7 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b69fb7645584c17b992960f611190bd : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb35216b42744a9b50d7c7ec4e2f97b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abdaef8d0de24f5b8e4e5601ab5999c9 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e7de1302bb44d3a8f52ddd52d95e3de : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0c32585a8e475695586d443e133ed4 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d3fb9c7eabe4fd4a62ec12334959dca : SUCCESS in 4m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/65e682d5dd564869bedeeb89b9dc45a1 : SUCCESS in 45m 54s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1dac26d9354b451a8217add6d71dd973 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5157063d5669297b0f4b8dd97cad8b2dd84f305f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:19:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"398889ada87e15a046f30188008523b8189a8bc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f598649b3dad41628727fb6a1713cedc : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41bd2feb595c410396f54fc1b3a3ced4 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10aae875519d4e6c96c44ddc2426f8e8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4935706561c91c1a9d603dfe41d3e004ac0d635a","revisions":{"e5e1a260246ee2a27fae66ef72b37aed3fe7bc0d":{"kind":"REWORK","_number":1,"created":"2020-10-12 02:54:18.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/72/757372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/72/757372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/72/757372/1"}}},"commit":{"parents":[{"commit":"38189e16709795537731c4089709185ed30e53ba","subject":"Remove Babel requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/38189e16709795537731c4089709185ed30e53ba"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 02:53:00.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 02:54:04.000000000","tz":480},"subject":"bump py37 to py38 in tox.ini","message":"bump py37 to py38 in tox.ini\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nref:\n  https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: Idc9b2ef918cf15472189c4c889752d98d9857be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e5e1a260246ee2a27fae66ef72b37aed3fe7bc0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e5e1a260246ee2a27fae66ef72b37aed3fe7bc0d"}]},"branch":"refs/heads/master"},"4935706561c91c1a9d603dfe41d3e004ac0d635a":{"kind":"REWORK","_number":2,"created":"2020-10-12 08:23:16.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/72/757372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/72/757372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/72/757372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/72/757372/2"}}},"commit":{"parents":[{"commit":"38189e16709795537731c4089709185ed30e53ba","subject":"Remove Babel requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/38189e16709795537731c4089709185ed30e53ba"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 02:53:00.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 08:22:55.000000000","tz":480},"subject":"bump py37 to py38 in tox.ini","message":"bump py37 to py38 in tox.ini\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nref:\n  https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: Idc9b2ef918cf15472189c4c889752d98d9857be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4935706561c91c1a9d603dfe41d3e004ac0d635a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4935706561c91c1a9d603dfe41d3e004ac0d635a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
