)]}'
{"id":"openstack%2Fblazar~757463","triplet_id":"openstack%2Fblazar~master~I7c7c2bd998d1a457ae1d19f42993e05370aa4e78","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"I7c7c2bd998d1a457ae1d19f42993e05370aa4e78","subject":"Reuse the docs deps to benefit from constraints","status":"MERGED","created":"2020-10-12 07:44:20.000000000","updated":"2020-10-12 09:35:23.000000000","submitted":"2020-10-12 09:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757463-1602495220027-0a98ea85","meta_rev_id":"b2ed0bd8570fc2f6129fe1099e2b4252598b5ece","_number":757463,"virtual_id_number":757463,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 09:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-12 08:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-12 08:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 08:43:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-12 09:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99e3e8cc56875357ebf660daee4e82748993efa0","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-12 07:44:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5f92b13b91e9e7056e248eb1542d6ee3c3798fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 08:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35bc4518a3d644d5bab4e98c31677959 : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e30bf89d0b364a858919dac3c769f77a : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f51448749daa4ec29462f172b6971072 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecfa4d979494a8ca463a36010eb847e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec3c7e0529404cc6b1446547d4818282 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b65777600aef4544960009bcd6542a24 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870b2ef9f69d47b7a6320e1ab973a5ab : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9489d1629124acdbce4fd4e97f3ee1d : SUCCESS in 3m 48s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ecd74c33455c42099be49f4f41c9f51c : SUCCESS in 48m 19s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/cabd83031c804b58805dbd9b50909d3a : SUCCESS in 42m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b561023ded1b4c7b83ca2cb7ceb26c3c : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbba5b569be5def01bb0b4af381dca07c9f1b4ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-12 08:43:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f43e7e07c346c90eb27d1da863b5ffb5c4e507f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 08:43:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b607dec99707964bc460681a605d824de0c32c51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:33:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9eb4f1d897174876b26dca31814e573d : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/129e6d0840254c0d84d0288fc4c4dcf2 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f423d2e1c364f87bba0c75a172093da : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f247cbe5426f40f0bcc67354eff622f5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad0d3fb92b69490289c1c05f2f9769fd : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28466c5bcbef4319a44e60bf3a243bcb : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/542c698736de4279a5c7da35dae491e4 : SUCCESS in 4m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/365e25f1fbf84b068d59aa639402b16e : SUCCESS in 45m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/2acdd5a6e4714eb7bbe7a8ad0857a5ee : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7a96a72b32f00e1fdedef57a8531d68e2c4c5feb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:33:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b2ed0bd8570fc2f6129fe1099e2b4252598b5ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:35:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a992eae706d04a179b3ebb6ba3142196 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5340e8f2fc4b4119a6ae7290cdbd9a40 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e504e8c1528644e5b6b9fedec8dfc046 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"184dbcc4a3d99895dd5c6cb4429c45d6a52c5105","revisions":{"184dbcc4a3d99895dd5c6cb4429c45d6a52c5105":{"kind":"REWORK","_number":1,"created":"2020-10-12 07:44:20.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/63/757463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/757463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/757463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/757463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/757463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/757463/1"}}},"commit":{"parents":[{"commit":"38189e16709795537731c4089709185ed30e53ba","subject":"Remove Babel requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/38189e16709795537731c4089709185ed30e53ba"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-10-12 07:42:21.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-10-12 07:42:21.000000000","tz":480},"subject":"Reuse the docs deps to benefit from constraints","message":"Reuse the docs deps to benefit from constraints\n\nTox trying to install latest versions for building releasenotes,\nwhich may not be supported by stable and lower branches,\nso should be restricted by respective version\u0027s upper-constraints.txt\nThis ps to reuse upper-constraints.txt from docs deps\n\nChange-Id: I7c7c2bd998d1a457ae1d19f42993e05370aa4e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/184dbcc4a3d99895dd5c6cb4429c45d6a52c5105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/184dbcc4a3d99895dd5c6cb4429c45d6a52c5105"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
