)]}'
{"id":"openstack%2Fblazar~761990","triplet_id":"openstack%2Fblazar~master~I2766db82ff18af9cd7eeefe237ca8ea329e11ccd","project":"openstack/blazar","branch":"master","topic":"remove-dead-db-utils","hashtags":[],"change_id":"I2766db82ff18af9cd7eeefe237ca8ea329e11ccd","subject":"Remove unused DB utility functions","status":"MERGED","created":"2020-11-09 19:34:50.000000000","updated":"2020-11-19 17:06:47.000000000","submitted":"2020-11-19 17:05:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761990-1605805500795-eaa21e4c","meta_rev_id":"0764aa71b8e7ed7c29dc1db4ddc578a562066e98","_number":761990,"virtual_id_number":761990,"owner":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 17:05:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-19 16:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-19 16:17:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 16:17:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-11-19 17:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b4f2f823f9bb7fc88abacd6bcc62a96f16f6cd6","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-11-09 19:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79f86c564c5d3385ff348e3462f603adbd2f80df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 21:01:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a0f5086bbf4454e9bde0476ef96e54f : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00ad24663b644bf7b5b641474b0fb398 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb331daa7464868a8498ed796ca2a61 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc9387311a945b886d25b95dac0356a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9821c96905c4e2f941c041c7066782b : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab52defe66a4a10b6fbc0dcca5396b4 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa24b83d5ba84a81b90fe1b6a6aec4d6 : SUCCESS in 6m 28s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d4bdba5080a949b0a5985563cf6d38d5 : SUCCESS in 44m 09s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/15f3983da11a45edb9c07121606c19f3 : SUCCESS in 1h 18m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6cabeed7433242ba8e827701361fff3b : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c5c30159682c4eac2c88163bda31ba33365787a","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-11-09 21:10:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"137b77f2bdac7c4cdfa7ceff3c3bc85447233631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 22:05:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cf74226c20e445fb8858f88ccffb2b3 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f33471b909646929d257fb6d4649231 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ac06ed5ee44611963710ab6a03bb4c : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58db8e9dd37245b589c5a517a8a5109a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fae88bda35104a1eb87ae0bbd51daf93 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f46534256204ba988f1be9385f00ce2 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fe2a2d440643a5a4e8700fce4f1032 : SUCCESS in 7m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b30b7d1586b546f480c68ea4900e52d9 : SUCCESS in 45m 29s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/462e72401b7248b29bd089caf1bb467a : SUCCESS in 42m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d01e9ec9acc4ba18ba81f79040d1e07 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04640dfc05b1dc8db423daa97b4650d9af079b90","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-19 16:17:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"145107adeb2d596ad992526f957505f769c57c8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 16:17:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4dc2fca460e578b02ec2a6549ab05de86c0b8b93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:05:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2881e43e422ad258c35cb6be960760d8fc7e19a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:05:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11776969b76346efa3b2cd139db68c38 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1988a6812e6433685dc0b1895874f89 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2af2862297421282d95d7f8f38bf5c : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c5352f7710440cc82f4af7ec819b299 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5169b76dec4b481492eb746a321ee39b : SUCCESS in 7m 20s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/71f18f290f0c4c51b5faa24a94b286f9 : SUCCESS in 43m 29s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/78363a9aaac74ced9ac54be7da323ea9 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e798c4847f6e66e71d548a77cbf92693c437fd50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:06:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65459de94e3b44f8b0ea14b255aaf68f : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0847f7e98b9f0359f9d0cd7675bc4cb94ab9fdd4","revisions":{"5ffb37056c5ab95008b91cd96c3f8129c55ea0a8":{"kind":"REWORK","_number":1,"created":"2020-11-09 19:34:50.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/90/761990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/90/761990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/90/761990/1"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-09 19:28:56.000000000","tz":-360},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-09 19:33:25.000000000","tz":-360},"subject":"Remove unused DB utility functions","message":"Remove unused DB utility functions\n\nThere are several DB functions that are not referenced anywhere but in\ntests. It\u0027s possible that there was some intent to surface them as some\nsort of reporting API, but that hasn\u0027t ever happened. Removing them as\ndead code.\n\nChange-Id: I2766db82ff18af9cd7eeefe237ca8ea329e11ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5ffb37056c5ab95008b91cd96c3f8129c55ea0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5ffb37056c5ab95008b91cd96c3f8129c55ea0a8"}]},"branch":"refs/heads/master"},"0847f7e98b9f0359f9d0cd7675bc4cb94ab9fdd4":{"kind":"REWORK","_number":2,"created":"2020-11-09 21:10:11.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/90/761990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/90/761990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/90/761990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/90/761990/2"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-09 19:28:56.000000000","tz":-360},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-09 21:09:51.000000000","tz":-360},"subject":"Remove unused DB utility functions","message":"Remove unused DB utility functions\n\nThere are several DB functions that are not referenced anywhere but in\ntests. It\u0027s possible that there was some intent to surface them as some\nsort of reporting API, but that hasn\u0027t ever happened. Removing them as\ndead code.\n\nChange-Id: I2766db82ff18af9cd7eeefe237ca8ea329e11ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0847f7e98b9f0359f9d0cd7675bc4cb94ab9fdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0847f7e98b9f0359f9d0cd7675bc4cb94ab9fdd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
