)]}'
{"id":"openstack%2Fblazar~762006","triplet_id":"openstack%2Fblazar~master~I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde","project":"openstack/blazar","branch":"master","topic":"shuffle-hosts","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-02-24 09:00:25.000000000","reason":"Change was submitted"},"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2023-02-24 09:00:25.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-02-24 07:09:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2022-11-23 11:07:34.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde","subject":"Add randomness to physical host selection","status":"MERGED","created":"2020-11-09 22:06:16.000000000","updated":"2023-02-24 09:01:28.000000000","submitted":"2023-02-24 09:00:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762006-shuffle-hosts","meta_rev_id":"81dda1b36baeb7afd79b240ddeffd72db6605694","_number":762006,"virtual_id_number":762006,"owner":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 09:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-24 07:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-24 07:12:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-09 22:06:16.000000000","updated_by":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"reviewer":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"state":"REVIEWER"},{"updated":"2020-11-09 22:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 16:14:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-09-26 13:16:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"state":"CC"},{"updated":"2022-09-26 18:13:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-09-26 18:13:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-30 15:41:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"}],"messages":[{"id":"13c2ea583d97392131d0ca856721a1913362861e","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2020-11-09 22:06:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e7d94056f9492610bb82cedbd3a0b9098e529f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 22:50:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e111c15c622d4416ac4affb29393b33a : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c282f70387d4dd090530a67796d0e6c : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c78d12e4974bc9b40188ec6122b6cf : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/773f96d169ae43b9a4313419d3bcfe77 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e43eb841d36644beb68beefddd213e8a : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0154966cd0134bb18e1c0b3070bf8793 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013bd2ea55014ae49b32583fdd7b7f09 : SUCCESS in 6m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0074b558fe714d298c5c2028d1ac20a1 : SUCCESS in 40m 16s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/641aca561d1146b09caac4cf4f731bd8 : SUCCESS in 40m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88ed7e7ef8f9492ea9d07039f7025bcc : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"632572f6dc4576c10f5a7409a4945181804acfe8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-19 16:14:43.000000000","message":"Patch Set 1:\n\nI believe I had suggested a change for this patch a while ago via IRC: adding an option to enable/disable the shuffle. IIRC placement does something similar.","accounts_in_message":[],"_revision_number":1},{"id":"dad9e55e701c776949487c52567d4af96649437a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-05 08:37:05.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease see previous comment.","accounts_in_message":[],"_revision_number":1},{"id":"e3de5651aeaecbe4f9ea03fe9709edcde08df6ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-03 16:49:01.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I believe I had suggested a change for this patch a while ago via IRC: adding an option to enable/disable the shuffle. IIRC placement does something similar.\n\nhttps://docs.openstack.org/placement/latest/configuration/config.html#placement.randomize_allocation_candidates","accounts_in_message":[],"_revision_number":1},{"id":"626fc8252b8c82761351ee8ff256d70734749d78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-26 13:16:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a538e825cfa294c98bec9866640e91bd57481b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:09:39.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39122816791142ed89627c9d03f24fae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee8674324904d22b7cdab0e0e2cf0a0 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd287e4f0e84733827a02cb78031bac : FAILURE in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b55258d0fd7d4fe79ddeddbdb98804bc : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc7cb1b7de7943bd82cb1cb5e8abd60e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1c6466bd054297b423644aa40a2927 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e78665a01804bedb87f173366bc0e01 : SUCCESS in 3m 28s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/391d8469552144639b291cc271bcfb99 : SUCCESS in 48m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/97ad0c9d8e754d3c8b9faae16515b55c : SUCCESS in 44m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b446fba00ebe4139a1531ce22121fe65 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d683e0c5d2c92353c94b7075fc47ba86a51f30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-26 14:13:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34ebeef99570def1096e6d7f20e9a0470d70c62a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:57:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bfce4c2267d43fca15d26b43b834fa1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a62468fbeec3454884ed7a1fc377d7a0 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca11d4110d74ebfb6463681d130af27 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91896969d6ee4b95935a274afdef5f98 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f722eff376c849b694164220bc192872 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2caf3a4084854184916bc76391bf5701 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3480f33d69574a6fbe89c58489520818 : SUCCESS in 4m 59s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/48ebcf44dc2343239cf0980944dc8ed4 : SUCCESS in 42m 40s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/150860f6148d48838f687fe53ba9994d : SUCCESS in 32m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd06378d185641de99b02cee0ce07271 : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ab4b45db4f4ea9a116eaf109bcf8263cd235054","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-26 18:13:09.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"65d6a300cfe135da311d064290a42827c6795a99","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-26 18:13:16.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"80159ce1bae92fb52e79097a9e4663939ddd8967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-27 09:55:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8b4031be8a75befc83f284a366c69d37a8baff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-27 09:56:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"25fc108c161e3b9f8ad59063840413da5d47e786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 10:49:42.000000000","message":"Patch Set 5: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a626d200f554aa7a3f8cacda0bce99f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd850c600af84717a437d5980beba141 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1f769478c54bd7b6b8e8893235f116 : FAILURE in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818a670954ec4528a98e67914790c3a7 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2d00f28f2b34e85829405cebb9e723d : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff9d240f356465e99afb2bb21b590b9 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cacd956335e14f4880ec3644d5a81ece : SUCCESS in 7m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/84ad20e8c2e740c4a1e5247ec97e6c92 : SUCCESS in 33m 28s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e9c29cb28642470abfd639161c4db537 : SUCCESS in 50m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e3adad6f0278491381bfe842d68574de : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd8e8c0a9088c03aff1ad1c83ff3cad1f75526a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-27 11:41:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b4d68219c4210cc94546ffee691176779b0662e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 12:32:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91aedc4a72942d293683c6c87f1761a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46a78d796f74703af994aec52cb24bf : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216be2d19779421da03e4fd51278e484 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcb637779d8649de85c6f6c26a7a64d5 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d723eceab0ed4914b90628a9a38b4775 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e56eac0a913489faad2e3f51eeac382 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b440aea225e649599904a3a17350b701 : SUCCESS in 4m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/db05fa6e5ae34a9286f7873e3d4e4c7a : SUCCESS in 48m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e73eb8a12f2c44ae974fbc3584e10562 : SUCCESS in 42m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb1c91ddf70d4dbd8972882ea3cc132c : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"92e8205456207bf67c3e9306e2d4447a4508285c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-30 15:41:17.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ac651782570649631133e2cd7f6df1a3203047b4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-11 11:44:20.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f2d0fc4112754f36300fad6d9a215e54a3fbfa4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-17 09:11:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b93976137587eacd15b8272b74d15fdef2b14918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-17 09:23:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8934df8bbc4afd84df4c42fb9f30df5b081b5271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 10:07:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dff24a462544870a787471403519ca1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b229cf739bd40aa95f0d99bc10f5552 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7f13c6633241759b9ee50d25e1b3af : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f830d7572fbc4575ad6e49a4acdc5c79 : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb6bbbfdb5ca4b73b71d4bb4148280b4 : FAILURE in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd34ab46d57a41dba310aac5c61ac0b1 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbf5b87eef1d4ed1b1bf5be3e0de5209 : SUCCESS in 3m 11s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/907d3a5d61414398a26091ca56a6984f : SUCCESS in 39m 42s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/35ba90a08551440eb09c35724f33e537 : SUCCESS in 38m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5abae96bcdb4f4ebb58ce4f80d045ca : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"810c775c78bbde464c96988a6ab452f4e5f63e34","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-17 10:30:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ecfd94178aa0dce74bff2f1d2f7eed68a9481c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 11:21:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/691875ed12b7400e92f75386515dc3c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cec0fa97059f4e0f939248ec04a9057b : FAILURE in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8014a71cda3416dbdf2b201223f9e9f : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd8795fedf8f4a048177ce6c4f06bb48 : FAILURE in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60b3b08f84d54802bdf52ea58691d207 : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e104c39cbe2425a9afce4077d15f172 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842c29cb77534fb4a5224f806b880575 : SUCCESS in 3m 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c1f22a94dadc418893581f87584398f4 : SUCCESS in 30m 50s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/eb607394c95a408eba12466bd1202308 : SUCCESS in 45m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7df415433cf4256920aa839f8a68033 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"78d341b10c9d7331381e9535004a809fd3afb139","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-17 12:19:45.000000000","message":"Patch Set 8: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6c388254e4a56459efba10af79ba4f10bb6922ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-17 13:22:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3d6e2dd625504850547bf19ee78a888f2dc903c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 14:02:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a689150c99e4a75b90ae9254e125060\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45f6ea9f82084df9a01de8aabe74d4e1 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e141784784a4e3082c29494dc85598b : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f0ff87287464251b5501d8483477953 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29200bedcb3a40a281817287f634eaa0 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be3f92fa02b492f8885d9a022c1bb34 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43bd8d33aaf34e3981812684fbf19930 : SUCCESS in 4m 23s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/189257bcc14b456c8de0697bd9c424be : SUCCESS in 38m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/26d25a968f37453aa6ad660d043dfc10 : SUCCESS in 30m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5c980b9fe38443483eae8a7161e3bb7 : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"42b716b148d8e05c2c1a31c01a57b3d9bdb4fa97","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-11-23 11:07:34.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"41a09b931452a5a4f2eeaf2916e46464d70e1ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-24 07:08:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2f40cb2a0bfa5d1a5288e9259413f79f23638cbd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-24 07:09:47.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"b9ad198da904ab9795a6c497f332327bb42548fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-24 07:12:55.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"744a0721e8373c3d29ab1aec3badf7928f6dd6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 08:07:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8acd9a262245ff9f0da83ff695f12a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eda9b90443d943d0be422b6d533e68b8 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f641151e4e452ab569a9fb6852735f : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bca0ff26f27a4195bf01c8e892606738 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a74e4540213e453dafa40f4448946588 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea1fc19584c4ac4b49aa2a708a0aa2b : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1d5b9abba3b4eef9035747408d863f2 : SUCCESS in 4m 05s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e29bd2ed36c8496fa61180f7fcf2e83d : SUCCESS in 48m 08s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b0bbf7b86d5d49549a415efefce2bc04 : SUCCESS in 33m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4937b71848b049c4a51f8b432b5e10fb : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a8eff97308575f66f1bc8865193a3e8015e24c0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 08:08:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"5087d6e31411af3e8e98e4a03b8bd71600a32ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:00:25.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2f46b682c7147f6ae716f4521b1b426\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4ce38fe60a407a8d5eaebdf46585a3 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e009471f8dd4463a15db942c8f128e1 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b2eed42335e458fa7fa10f160a531e1 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5289ae92344f568283024750efbf5d : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc212c68b8934fa99dd7a5b705a8e01d : SUCCESS in 4m 17s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5cd5309db3e7471494c0b11c9943300f : SUCCESS in 34m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/986874680b154fe1ae7a45a488aa7ab1 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":10},{"id":"67d78a92a4a0263961391b086a641c82a812003c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:00:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"81dda1b36baeb7afd79b240ddeffd72db6605694","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:01:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/333a3ed68fd343c1880c03212bd027d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86d46c3810294f42b71e69529f07236e : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a787939a317743f6af4a139fced6d7c6 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c4a43053749b172197d2c5aa527858a518a99f08","revisions":{"b8169dbf97098fc8766536599cbee6d0f0295ac7":{"kind":"REWORK","_number":1,"created":"2020-11-09 22:06:16.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/06/762006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/1"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2020-11-09 22:05:49.000000000","tz":-360},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b8169dbf97098fc8766536599cbee6d0f0295ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b8169dbf97098fc8766536599cbee6d0f0295ac7"}]},"branch":"refs/heads/master"},"be15ed697777c6f3746b78ee380bb56b17799b91":{"kind":"REWORK","_number":2,"created":"2022-09-26 13:16:06.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/2"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-26 13:14:51.000000000","tz":0},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/be15ed697777c6f3746b78ee380bb56b17799b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/be15ed697777c6f3746b78ee380bb56b17799b91"}]},"branch":"refs/heads/master"},"f06cb54ee88149678b2943801ae0e521502f88ba":{"kind":"REWORK","_number":3,"created":"2022-09-26 14:13:54.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/3"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-26 14:13:32.000000000","tz":0},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f06cb54ee88149678b2943801ae0e521502f88ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f06cb54ee88149678b2943801ae0e521502f88ba"}]},"branch":"refs/heads/master"},"711c73a1b742b545462f02b0af642039a598b7bd":{"kind":"REWORK","_number":4,"created":"2022-09-27 09:55:21.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/4"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-27 09:55:09.000000000","tz":0},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/711c73a1b742b545462f02b0af642039a598b7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/711c73a1b742b545462f02b0af642039a598b7bd"}]},"branch":"refs/heads/master"},"830e1225e20e251241d76366ab35ed101be38da5":{"kind":"REWORK","_number":5,"created":"2022-09-27 09:56:36.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/5"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-27 09:56:26.000000000","tz":0},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/830e1225e20e251241d76366ab35ed101be38da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/830e1225e20e251241d76366ab35ed101be38da5"}]},"branch":"refs/heads/master"},"f0536a553150cd793dbd5a048455f149f73e6322":{"kind":"REWORK","_number":6,"created":"2022-09-27 11:41:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/6"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-27 11:41:21.000000000","tz":0},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0536a553150cd793dbd5a048455f149f73e6322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0536a553150cd793dbd5a048455f149f73e6322"}]},"branch":"refs/heads/master"},"6ccac0ff713b0a977b96192cadeb9eef0d5ff249":{"kind":"REWORK","_number":7,"created":"2022-10-17 09:11:27.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/7"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-10-17 09:09:06.000000000","tz":60},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6ccac0ff713b0a977b96192cadeb9eef0d5ff249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6ccac0ff713b0a977b96192cadeb9eef0d5ff249"}]},"branch":"refs/heads/master"},"e551189fc470cffcd0b8a0e9c0ca9d339fb0fd39":{"kind":"REWORK","_number":8,"created":"2022-10-17 09:23:10.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/8"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-10-17 09:23:02.000000000","tz":60},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e551189fc470cffcd0b8a0e9c0ca9d339fb0fd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e551189fc470cffcd0b8a0e9c0ca9d339fb0fd39"}]},"branch":"refs/heads/master"},"4165b66427f5fe13940ef8f7bf1dc67a6f67c4c4":{"kind":"REWORK","_number":9,"created":"2022-10-17 13:22:01.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/06/762006/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/9"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-10-17 13:21:51.000000000","tz":60},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4165b66427f5fe13940ef8f7bf1dc67a6f67c4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4165b66427f5fe13940ef8f7bf1dc67a6f67c4c4"}]},"branch":"refs/heads/master"},"c4a43053749b172197d2c5aa527858a518a99f08":{"kind":"REWORK","_number":10,"created":"2023-02-24 07:08:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/762006/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/762006/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/762006/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/762006/10"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2018-09-10 14:59:46.000000000","tz":-300},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-02-24 07:07:21.000000000","tz":60},"subject":"Add randomness to physical host selection","message":"Add randomness to physical host selection\n\nIn the event that a given physical node has reliability issues, it can\nbe difficult for users to \"get around\" the problem because we currently\ndeterministically pick the first node from the list of nodes available.\nAdding randomness to this process helps spread the probability that a\ngiven lease will nab a node that has issues (but has not yet been\nidentified as such.)\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I0f0ddb14ed5d21f0ccc5c9659c821d58cea5cbde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c4a43053749b172197d2c5aa527858a518a99f08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c4a43053749b172197d2c5aa527858a518a99f08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
