)]}'
{"id":"openstack%2Fblazar~762956","triplet_id":"openstack%2Fblazar~master~I95833549861a0c3500d558ca2ebb0986fadc2c06","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"I95833549861a0c3500d558ca2ebb0986fadc2c06","subject":"Update TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-11-17 06:04:43.000000000","updated":"2020-11-19 18:15:48.000000000","submitted":"2020-11-19 18:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762956-1605809633418-354c625b","meta_rev_id":"b48b22fee897766289fde0e35c6c8d13a5e285b4","_number":762956,"virtual_id_number":762956,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 18:13:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-19 16:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-19 16:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 16:10:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-11-19 18:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dae1b30d0480584db3f8a54e60a4725754059210","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-17 06:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9335a2b6e6d6d16d8f21c67e1aaed45a6c6a8e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 07:18:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/927898b7e1e446a8a7273536555f9c9c : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f01400a0c60547f6911ab17ad383f4ed : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4247bef628914265843a351c36d69cbe : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053e6e51a31943d8a7e0c516a7b69130 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1782572c734a50998eeb92cdca1284 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/881446bbcbb24630b2f94ec0f43c3a8c : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6efc92a1b9487998deb6695da2c4c1 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4bb24d55fd84ee5a575f291e844dc4f : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a489a7414d7405d97dee917eeb6c144 : SUCCESS in 4m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/830ad96dc66246d4948205673b11de2e : SUCCESS in 42m 37s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/26e2af968847488fbb607927e63af9ea : SUCCESS in 56m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b3495588a8a43ae978de3c2b786596e : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ec5ee517991a1cfbf4d2baca8c8534415fa4205","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-17 08:14:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb52dbf24c156f1ef07eabc0d03854bc089cf833","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-19 16:09:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cab063f32dd9afdc2fb326fd7500d8fbd7ef4e6f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-19 16:10:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"776370253fdfeab0a8926b3fdd7e9f2c244f5215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:28:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/10b8c60d72494b238f5b8fe63b47cf6a : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5512433304074917ad873c9c5be1e492 : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba0d933429134da194f435d63ab4113a : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aee0771380e4e1b99b13bc341aa0a28 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/785c68184e994a47b36ac74f7028060c : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9061df5bf5b44e9914bc40b2c4de400 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3c7325bde634fbb98403a41243e8363 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0423532c01432e91207b2e1b300f38 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3f932584dfc4328bb2599c55d3d27d2 : SUCCESS in 4m 17s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1c860cf9659a4d65a6aef4540f0d520f : SUCCESS in 58m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0923b6dab6d7420c97de98f2e64f77a1 : SUCCESS in 52m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88179f1975084098bab7e47be18dc4c1 : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d93ec9e460ab0f48ebc248cfbf41bdab06fab6c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:28:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"812228182960ebd55533b922ae7a394a49534eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:13:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/402600b2b6ef4c70a579590c2bb39aff : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/488f9dd5c1cc4cccae4a55b5b6e11056 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdff5725f66144dfacf912934587743e : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4f209112964fedb43594b31c901eba : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba69335e941241e289cb91bbd587d1a3 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1907aaab154e9bb3676092bdfd3d9f : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8521cbcef824a278ffd5be95e19a4d7 : SUCCESS in 4m 08s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/812c266389bd42e8b220615990c3b775 : SUCCESS in 41m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c544426a61284f458260c2e0cee60eb7 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"05acc01b3da8f2b62cadba4fbac4259991db6c3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:13:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b48b22fee897766289fde0e35c6c8d13a5e285b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:15:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ba217ba85a9541dabfbe7c7487e9d646 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a79ab6e3eec49d6a1dcd7e288d331b9 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f56aeca9fb74ddabfe4ffbdee2f334a : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ccf172708b0bff3d2642fc2d26fb25ec874e0e89","revisions":{"a17efd3a89ced74b758283fd07a3d353ca269964":{"kind":"REWORK","_number":1,"created":"2020-11-17 06:04:43.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/56/762956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/56/762956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/56/762956/1"}}},"commit":{"parents":[{"commit":"9008aae4494a3f6833b01127b8b40e3f0c77ed04","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9008aae4494a3f6833b01127b8b40e3f0c77ed04"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-17 06:03:20.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-17 06:03:20.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: I95833549861a0c3500d558ca2ebb0986fadc2c06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a17efd3a89ced74b758283fd07a3d353ca269964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a17efd3a89ced74b758283fd07a3d353ca269964"}]},"branch":"refs/heads/master"},"ccf172708b0bff3d2642fc2d26fb25ec874e0e89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-19 16:09:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/762956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/56/762956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/56/762956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/56/762956/2"}}},"commit":{"parents":[{"commit":"368403ea806e76405a773eb3cb679ff7c77334f4","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/368403ea806e76405a773eb3cb679ff7c77334f4"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-17 06:03:20.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-11-19 16:09:59.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: I95833549861a0c3500d558ca2ebb0986fadc2c06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ccf172708b0bff3d2642fc2d26fb25ec874e0e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ccf172708b0bff3d2642fc2d26fb25ec874e0e89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
