)]}'
{"id":"openstack%2Fblazar~766841","triplet_id":"openstack%2Fblazar~master~I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5","project":"openstack/blazar","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-12-13 00:12:14.000000000","updated":"2021-02-01 22:53:56.000000000","submitted":"2021-02-01 22:52:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766841-policy-json-to-yaml","meta_rev_id":"df653ec79751f1f22c7d3de56d54e28ccb7c6137","_number":766841,"virtual_id_number":766841,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 22:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-01 22:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-01 22:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-13 01:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 09:09:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-01-20 06:22:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"b02c97289a97559d9e4904357d099a36e53ca4e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-13 00:12:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8641681241ffa2c58365832c02e601a9afc40f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 01:09:51.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aff32035da3240f5a8bffe91de834b46 : SUCCESS in 2m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/878ffd075d40410a9fc1d358818526da : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cfcd4e353ad4ae2a8a9e421d27c8d97 : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e75b45cc9d1746588b026ae722b6dc59 : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519e064015494e629665830de3d8397b : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e544a23483904833be7c88f15d79ad42 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d19424f561c44d5ba7f698b51ee620ae : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d17071bf6646a68f4ef83307347912 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05dfaf0966824abe95bc9535314b5d7f : FAILURE in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eae3b5e4ffb74eddb70ddb5ffdb2f5d0 : SUCCESS in 3m 42s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/69ff03aa8662410f8de54b0f665ba170 : SUCCESS in 51m 33s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3682821d8180443d8fc39d8af2df46b1 : SUCCESS in 54m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11cefb40e7e34b8dba8bade54a9c9887 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8457d50f2b705e60877c4744e6808271e3f8298d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-13 01:14:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"724939d3c5aafa71a0a0d1b4bc7f13631d84bcc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 01:52:51.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/645311a25c9f4edabae277060b85f310 : SUCCESS in 1m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/76f2e5fb002549ca8ee1ee9ac2ddba55 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdbebb04a5ae4084a0cef1b2626c8fe7 : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43153180e28049a6b78ce03510a4760c : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4918f38f244e4d41bce1a3d4874da799 : FAILURE in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8064205ff3844db9b822f779c1d6d4e2 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/090a8b1dbce2409b9c006ca789e144f1 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b0dabcc0eb547678e6dfbb391201f4c : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47982a05a0964c6a904395f4d39f1c05 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784354c7773b49089a2124fa74a826e5 : SUCCESS in 3m 17s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/417a1441a20e4747aea8312163912768 : SUCCESS in 35m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a2ceb5726ed94002b48a9c2536638898 : SUCCESS in 37m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14f4897e52cf4cd699340eb70b5b1876 : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc6014e5cb6c77f5795e4394504dc0f8e258ec85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-14 23:50:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7ff396c4c4e76ec7be12edf5e4e835fe19dd2334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:31:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/afa06c9280a94ed5b45ec2c804f1f80e : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1657808d7d9e4767aef6f2e0ea9786b0 : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd04c3237c0c4e36aa32b96483c89586 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282b49bf15854a108407f24d319ca17c : FAILURE in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfa07834f8d416ab969f331b9ecd5f3 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e973f926f224a61aa59a9fa10061611 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a909ffb4378a4c9ba8e536f8208cc67b : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85eddaad3500499fa967ce1705dc9987 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bcf26efa5847f89616343a9dee7b78 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4092af8405545a48139f342969d4d20 : SUCCESS in 4m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/9cf83f34ea784d17afa5215c9a840cd5 : SUCCESS in 36m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6cde0796d3d147d2825fbb3514212773 : SUCCESS in 37m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/044dfefa26f0482f988e06606883b939 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05429e50627aa6e6e0a9970d442ba16c9b5510bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 09:09:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9be5570e96ba8e8a47104895f4575d81764d5d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:55:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e92d1b6852a4ce6b5ed36f4f2c22dca : SUCCESS in 2m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d278b6c61254d1d97eac4d8e018e647 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2758a73e097c4d50bb2966b4d7c2d709 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/baac97c41f124f74ac9f5d8954a41134 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1925464580412794770bf76b6706ed : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fee5791bfc8a402abff954cc3c2135ee : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/028607deff3e4a67954a43a72c812533 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33d4814380e4a9fbea80d31a57c922d : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0611359299fd487bb5bd1bc4b2a0994c : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/560d2d6e06d943bda91e0ecc86a88ac6 : SUCCESS in 3m 59s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1d787c90d1114d2f9757bcc0bdc5fab1 : SUCCESS in 40m 13s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/998a8c53a6334207a488b1db7bc500f9 : SUCCESS in 36m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce582d1a95a444229c08b4c6ce0d2b44 : FAILURE in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"55d0fd3cf25260a635919951423ee5624fe14e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 10:13:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c6e90a9aa746ffc32093508c0ce84bd447008037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:11:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3efc8ea1beef4030b946bcf7b980a115 : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f5472cfe9a54424ab21e64ffde6ade2 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/487559566a1544689060453bfcd9ab06 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7e37f9a7b954c2398ccffffc6d2b88a : FAILURE in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075143cd8a9845c486d99afc402a1f02 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bbd4e1b6f654ab881496016e46135f4 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e1d1464c084da5850b7b63c0bc77f9 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dad96920feb44570889d6cf1a967aca2 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd164854b7ba4481af1d402a1ec2776a : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed2568b6df5b4f868549e27655a61354 : SUCCESS in 7m 39s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8eb6b96e5a0f4f6abcf17ddf004cfcb4 : SUCCESS in 38m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e4cf3819ad23472fb0fa7248017a8d4a : SUCCESS in 44m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef7c80b41e9449fc8614c1a8cc34885a : FAILURE in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"60b51bcdb3514479690a03cc0b4ce988e7613862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 12:09:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c6af1e27798034489688068fa973702a6d9ed6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 13:30:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66689d40f72b4fc19dbecd243ae842e8 : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9eb1e5bfbcba4edd8757d2371f574dd5 : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":6},{"id":"649048ed990b6bccb2872390b7540a54d98ea274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 13:35:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3cd8beba118171232a5e983810bf1c8a835f1f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 13:49:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d87673d8c0ef4b1da53c4abde777d1df : SUCCESS in 1m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03bd7418a5e4422c92ae9e162f216f26 : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":7},{"id":"02643eb96d1c695816acbfae46551c683034e515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-12 13:55:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9a490922112072cabe9a1877a155f92d348bc05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 14:26:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84e7319a6084435ab1ba7c91f4725bd1 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fd71271c6b14daf8d47b872ed23c868 : FAILURE in 25m 14s","accounts_in_message":[],"_revision_number":8},{"id":"37a12c96ce981567c158fb0e09c2a9c5071f35a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-14 23:00:34.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"b1694983c8a1e0d5c4fcbf5da5c7c572df073576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:09:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe30d7899bc344a1a7fb31338a05cf8a : SUCCESS in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19aec81fab9b462b9fa74d95b05cac53 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":9},{"id":"af458dd604e556aeb1305966594d2de49c36b39f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-20 06:22:46.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5bc67c242aa3ac6c5011df6dd2130d623cce87a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 06:47:44.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb6c1185455b4ab294cea02bf1d12682 : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cd5504f9e194d328ac9cc2a29e01f55 : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":9},{"id":"237aaa8c597f84ae51038cb3bda3ff3b4d90d0b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-28 15:13:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"83b210a7ac5e2255dff2b680e1936fa1a503599b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 15:20:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fab05b6023c24aeb91c99c056170594f : SUCCESS in 2m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e95df98d9084e03ba83dfd3649db76d : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":10},{"id":"e48e5edc9266c55813420dfaccb20f96cfa11188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-28 15:26:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"afc50b5ad6c1d7d6a849aafce2d02cf4226ad053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 15:35:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/708f5ae8223f41a592d1b5b92608f7eb : SUCCESS in 2m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8ca5a3f26f3431ab47cd4282df0acaa : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":11},{"id":"259e2959ca90819743efab4b687de6397c67c4d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-28 15:37:49.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"430f1667285ce0a05b63eb2ad4960eda4bb1a592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:32:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ae68b3af9bd4fb3a7e9032999d7269e : SUCCESS in 2m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4459b826d3464e56a48b4f3b89753b95 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ea2a9a876a44fa3baca22c3d07bbef9 : SUCCESS in 13m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07ea30336067402083bca7cd061a1163 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a27500d92d04534b1d0ead9d08c5ba6 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f5a972b73d4bb1b9b514ac44e5e0b2 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0f86e083f4941cfb5b4f558d8f7acb0 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2856358832104661bc745fab92c7a7d6 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461a7ca54c4a4596aa9cefce273a5ccf : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bb4af5449924490a785136e7076e5e5 : SUCCESS in 9m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/83aededd5b7f48d08d84119127af7028 : SUCCESS in 43m 12s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ae3b713c139540c5b7143e78a8d2b953 : SUCCESS in 44m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba4a52d2a6db43c9aee2f8fd2da13e81 : FAILURE in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ffd77ca00c57eb1c628c980454f061ede312b58d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-01 17:36:07.000000000","message":"Patch Set 12:\n\ncan we merge this as it is all good to go now?","accounts_in_message":[],"_revision_number":12},{"id":"7834394963013bccbd82d4d6ea30ac27aab17c7e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-01 22:01:58.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"90935219abe687d63aabccc1791fdf491c418e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:02:13.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"f4d3947b3df4f9980f9fe48bf3b71629d95cb1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:52:01.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bcff89ed7d62449b9de766fe216e5833 : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03408f60abfb443eb71fe143a7ca8e17 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d18a28e07f4e6eb3bc176bb8314365 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3b521caf5c4f60b5c065320c5ed98b : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/582567f7fce044e8855ba6bb641b69d3 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eb224edfd8247a9825f0ebfac4af2a3 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cf92196b924499af49a89ab7434a2f : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4960fc1d43e458f8ac15f08380c588c : SUCCESS in 4m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ae9e20cd25454af7ba1a780b7285c652 : SUCCESS in 43m 33s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/52fe9798270f464ab7c082f9586f431d : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":12},{"id":"fdbf78ea4e3b364b56222223872d9cbf43b31def","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:52:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"df653ec79751f1f22c7d3de56d54e28ccb7c6137","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:53:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda4e65896be402798148b901bd0fbcb : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd7567604cdc47d19b37de21860ce10a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"3f1c1ed3125445387309e59cfe243c824fbe7853","revisions":{"100f18b8052583fa9cba935306a99d1a1c4993f1":{"kind":"REWORK","_number":1,"created":"2020-12-13 00:12:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/766841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/1"}}},"commit":{"parents":[{"commit":"cb7c142a890e84a2b3171395832d9839b2d66a63","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cb7c142a890e84a2b3171395832d9839b2d66a63"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:12:13.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/100f18b8052583fa9cba935306a99d1a1c4993f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/100f18b8052583fa9cba935306a99d1a1c4993f1"}]},"branch":"refs/heads/master"},"35bb9de688ec91a1b68a7fea826cdd028c0fe2c2":{"kind":"REWORK","_number":2,"created":"2020-12-13 01:14:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/766841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/2"}}},"commit":{"parents":[{"commit":"cb7c142a890e84a2b3171395832d9839b2d66a63","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cb7c142a890e84a2b3171395832d9839b2d66a63"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:14:44.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/35bb9de688ec91a1b68a7fea826cdd028c0fe2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/35bb9de688ec91a1b68a7fea826cdd028c0fe2c2"}]},"branch":"refs/heads/master"},"0a1b064aac1533d1b632a055ee729fcdca1999e6":{"kind":"REWORK","_number":3,"created":"2020-12-14 23:50:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/766841/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/3"}}},"commit":{"parents":[{"commit":"cb7c142a890e84a2b3171395832d9839b2d66a63","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cb7c142a890e84a2b3171395832d9839b2d66a63"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-14 23:50:49.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0a1b064aac1533d1b632a055ee729fcdca1999e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0a1b064aac1533d1b632a055ee729fcdca1999e6"}]},"branch":"refs/heads/master"},"e81dd05ffea02d105bd06187427636edc2fd794b":{"kind":"REWORK","_number":4,"created":"2021-01-12 09:09:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/4"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 09:09:35.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e81dd05ffea02d105bd06187427636edc2fd794b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e81dd05ffea02d105bd06187427636edc2fd794b"}]},"branch":"refs/heads/master"},"1f486c27f22e912ae76e7d801666b6d0e5227f47":{"kind":"REWORK","_number":5,"created":"2021-01-12 10:13:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/5"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 10:13:46.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1f486c27f22e912ae76e7d801666b6d0e5227f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1f486c27f22e912ae76e7d801666b6d0e5227f47"}]},"branch":"refs/heads/master"},"c493bcbc073fc528adea360348b32b723efa3188":{"kind":"REWORK","_number":6,"created":"2021-01-12 12:09:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/6"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 12:09:33.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c493bcbc073fc528adea360348b32b723efa3188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c493bcbc073fc528adea360348b32b723efa3188"}]},"branch":"refs/heads/master"},"ae1c6c6de96d960f96b358f6be722be0e9edcfd7":{"kind":"REWORK","_number":7,"created":"2021-01-12 13:35:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/7"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 13:35:48.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae1c6c6de96d960f96b358f6be722be0e9edcfd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ae1c6c6de96d960f96b358f6be722be0e9edcfd7"}]},"branch":"refs/heads/master"},"7a9360f319a8a00982de1fc35abc2ee696832078":{"kind":"REWORK","_number":8,"created":"2021-01-12 13:55:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/8"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-12 13:54:56.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7a9360f319a8a00982de1fc35abc2ee696832078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7a9360f319a8a00982de1fc35abc2ee696832078"}]},"branch":"refs/heads/master"},"ea800ee68dd8191bbfe717ec1757041d5ce38e54":{"kind":"REWORK","_number":9,"created":"2021-01-14 23:00:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/766841/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/9"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-14 23:00:31.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ea800ee68dd8191bbfe717ec1757041d5ce38e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ea800ee68dd8191bbfe717ec1757041d5ce38e54"}]},"branch":"refs/heads/master"},"cb75185f820c83fade66ddbd21d7aeec725e104f":{"kind":"REWORK","_number":10,"created":"2021-01-28 15:13:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/10"}}},"commit":{"parents":[{"commit":"7d131c96d11799a463e8e8e0d15251175fb48ea1","subject":"Re-use rpc transport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7d131c96d11799a463e8e8e0d15251175fb48ea1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 15:13:43.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cb75185f820c83fade66ddbd21d7aeec725e104f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cb75185f820c83fade66ddbd21d7aeec725e104f"}]},"branch":"refs/heads/master"},"a47de2df742b77ab98167113105f2fcb849b9084":{"kind":"REWORK","_number":11,"created":"2021-01-28 15:26:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/11"}}},"commit":{"parents":[{"commit":"7d131c96d11799a463e8e8e0d15251175fb48ea1","subject":"Re-use rpc transport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7d131c96d11799a463e8e8e0d15251175fb48ea1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 15:23:35.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a47de2df742b77ab98167113105f2fcb849b9084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a47de2df742b77ab98167113105f2fcb849b9084"}]},"branch":"refs/heads/master"},"3f1c1ed3125445387309e59cfe243c824fbe7853":{"kind":"REWORK","_number":12,"created":"2021-01-28 15:37:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/766841/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/41/766841/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/41/766841/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/41/766841/12"}}},"commit":{"parents":[{"commit":"7d131c96d11799a463e8e8e0d15251175fb48ea1","subject":"Re-use rpc transport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7d131c96d11799a463e8e8e0d15251175fb48ea1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 00:11:07.000000000","tz":-360},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 15:37:30.000000000","tz":60},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file format from JSON\nto YAML [1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027 config option\n   from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with upgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side via\n   warning in documentation and release notes.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I9ca5459a6968d93dee1bf24b3e0012a4cc7020b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3f1c1ed3125445387309e59cfe243c824fbe7853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3f1c1ed3125445387309e59cfe243c824fbe7853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
