)]}'
{"id":"openstack%2Fblazar~770252","triplet_id":"openstack%2Fblazar~master~I2e6e9c3b3d934e1ba60b07d239e970137c32a48f","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"I2e6e9c3b3d934e1ba60b07d239e970137c32a48f","subject":"remove unicode from code","status":"MERGED","created":"2021-01-12 09:22:32.000000000","updated":"2021-02-23 11:40:02.000000000","submitted":"2021-02-23 11:38:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770252","meta_rev_id":"212e51faad2432af4a9de245f315959d801e6594","_number":770252,"virtual_id_number":770252,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 11:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-23 10:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-23 10:57:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 10:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 10:57:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"59894a6135d7d34091d11b3ba8f68966fe56cb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-12 09:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1607e6ed048b80416426dc7f22c8dd66be25df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:17:28.000000000","message":"Patch Set 1: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab104be68b7649e888503f30b609121e : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc2e900b0b5a4a1988d0129dc45385f9 : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48edf5b25615492aa79617ceb1c169e6 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf76e7e4cea4e12877fe3d4db38082a : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b66924d555964f2f83c72aeaa1b5791f : SUCCESS in 22m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e6c12d7a9347beb4770c593d0ad0f9 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed2d372c488c4155b458cd6a799902d7 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c1525af10b42ffb308fd1cf7d3ad3d : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e3564dd0a241129464e7481a119ffa : SUCCESS in 4m 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e8e0cf07addd447ca44ecba855b159af : SUCCESS in 45m 40s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a061194d011e4212ac34eca20a682595 : SUCCESS in 42m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dda8975064fc4d40a83278c8ccac2f2a : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a73684c301bb8ec7435da8da9a895d8d6e562b7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-12 12:02:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd252dd93326a348b4800ad9ca674def67a9f984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 13:19:59.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62efc7c749494ac4a26852eaf46e5996 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7185bb9624854740a1248c94e34af91d : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8bce93a8ac4439bbbb4d5c89c6c9f7 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c25302b1bc84ec5874f355cceea8900 : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f5a1a8ddac9416f9d34eaeb5007920d : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b244c9d85b0049dcb46a1e159f0641f4 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83d4dd860b0a4828b57a3df7a29db87e : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14692533101a496b9a4684f6538ebebe : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a73aa3b1f36346f585ac3893e32788ab : SUCCESS in 3m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d0d145eb72184e98aeef577f05727970 : SUCCESS in 51m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/614019f81bf646c0aad1a832cb74c101 : SUCCESS in 37m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d158b24d6dc4f3caec7b1e11dd94de7 : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26f7613081bfbca2c99c27bdab148ed5c49c853e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-13 01:26:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6b56334279c7342039e27f06ed1c2983121d30e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 02:30:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6985635c2b15453f89479c6673efbe02 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/557bc70ae3c64347b72c6a69648a8b5d : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb06287b0e7461889a92ae51fb05b4b : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0aa20e6d8e4a9e95f0e19f3df9269a : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc016f91be244b9c842681aa0188af7c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f208bf992fe34b1096adff105cd68203 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/960e9077e99f415aae4ca7b77805f8f4 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a8b33629c44c2bbd4230a30dec2e98 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a350db1a5c1848afb3e775968d28cb7a : SUCCESS in 3m 35s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e561d685852f4dde8221ccb03ae09bc4 : SUCCESS in 50m 21s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/cadcf5a5943049baa72013add9e4fc14 : SUCCESS in 42m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b79d1505166b43cfa6caea7da9725342 : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cd79d066e86f06fd585981352d7dd58ee237819","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 10:57:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1e0732fb7d6ea2b2fb822ffbdecb98e882836fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:58:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63907624f3001f7e409a51e269cc464ccee425e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:37:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fc4ae8f3be7c480d8ab87ff0193511c0 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a119a5eab541dd8fa740a4b83fe158 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782598d3343f47d3a4f57748ddd0312c : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eaa607a97a24ea09e0543240ce0ca97 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71327bc6964e4df1ae53e8d7a3d5e9ce : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836970f997f441279755b0043994f32a : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a07b84e64314c6e9ac628023e254a79 : SUCCESS in 5m 52s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/412384d369384d29b423f2cb2f305b45 : SUCCESS in 37m 12s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/56194d50490a4ed59727be05e2d5f416 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f6ef6e054d305129f3add4cc8826abda982dd4f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:38:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"212e51faad2432af4a9de245f315959d801e6594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:40:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0a3a753f8a0c43de812d14229ff9768a : SUCCESS in 1m 03s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007e78b059c247c8a54c5b0be74092ca : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2914e9787d44199225bbd85acda2d9 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e81bbc72b7985fce4c4d903ffe2a9a0a61c8d823","revisions":{"5e2ab3ab576436fc27c09d534dd52ea159b9906a":{"kind":"REWORK","_number":1,"created":"2021-01-12 09:22:32.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/52/770252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/52/770252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/52/770252/1"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:18:50.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:22:17.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I2e6e9c3b3d934e1ba60b07d239e970137c32a48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5e2ab3ab576436fc27c09d534dd52ea159b9906a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5e2ab3ab576436fc27c09d534dd52ea159b9906a"}]},"branch":"refs/heads/master"},"2ae34d5db29c56f4a5edd245079dfeebbc8fd552":{"kind":"REWORK","_number":2,"created":"2021-01-12 12:02:53.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/52/770252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/52/770252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/52/770252/2"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:18:50.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 12:02:36.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I2e6e9c3b3d934e1ba60b07d239e970137c32a48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2ae34d5db29c56f4a5edd245079dfeebbc8fd552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2ae34d5db29c56f4a5edd245079dfeebbc8fd552"}]},"branch":"refs/heads/master"},"e81bbc72b7985fce4c4d903ffe2a9a0a61c8d823":{"kind":"REWORK","_number":3,"created":"2021-01-13 01:26:56.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/52/770252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/52/770252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/52/770252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/52/770252/3"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:18:50.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-13 01:26:38.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I2e6e9c3b3d934e1ba60b07d239e970137c32a48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e81bbc72b7985fce4c4d903ffe2a9a0a61c8d823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e81bbc72b7985fce4c4d903ffe2a9a0a61c8d823"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
