)]}'
{"id":"openstack%2Fblazar~771576","triplet_id":"openstack%2Fblazar~master~I55fbae296f5582148f5b48b7cf96fbd77d014227","project":"openstack/blazar","branch":"master","hashtags":[],"change_id":"I55fbae296f5582148f5b48b7cf96fbd77d014227","subject":"Make lower-constraints job non-voting","status":"MERGED","created":"2021-01-20 09:00:24.000000000","updated":"2021-01-20 22:06:57.000000000","submitted":"2021-01-20 22:00:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771576","meta_rev_id":"6ef16f95afbdb184a647baecd24406ab9e788932","_number":771576,"virtual_id_number":771576,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 22:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 21:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-20 21:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 09:43:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"661e03cf0c0b30706f09ef4479854245b5bf28d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 09:00:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad96673b9d23837e8d8da9ba8f9e472ff319dc2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 09:43:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0115fdf9b8314ba882dda0a5f5fdc0c5 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7871f049839441583428a10275134ad : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8dc9f6845c247f3bc6414ad26ded717 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54ea9772f701472bbfc49d2950be4ad7 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e46f93edbd44b48ab8a0736dc6ebabd : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c19b797c91f40ba8488940c8d9b73d3 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b72a6768a8f41de8a17cfb39258a660 : SUCCESS in 6m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5c58eef2a6dc42dcb5d075df2d58be5a : FAILURE in 34m 13s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/23693176a65f420aab0db923120e90e1 : FAILURE in 26m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e534f70aeb34301aab49ba8fbb55555 : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bb13e7ba9246cacda414dee6290b757ff9d14bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 14:06:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb06b94c50832ff5cbb7ab5da171008c0bda6e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 14:38:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd2fb2b52554ed1b1800fb18e4b99c2 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb99adb6ae26485589929afb2ebeb84a : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d68bee21da46498c0c9c183666a900 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6af27e81d7849d2b5c8ceed54a2289f : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0485a448fd3435bbb95f64b328fa828 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d2621f5650d4ada99bdf12380893200 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dacb6ea12534c85a7f0cb1e52284541 : SUCCESS in 8m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a65790cfacca4c6bafdff19722cbf0c1 : FAILURE in 25m 30s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4001f7fda5a24e85a6a8d63a28d56c0c : FAILURE in 26m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7dd887e677b4cf3b0914bf583da2096 : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9304dae4880ed31a61f3cfed842a68d315e27e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 20:06:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"279a4e923708ab055f3428094ae83486eb56ad2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 21:11:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e71a3b0e04c749218f3a5a3c80a6a142 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba620b81c66c4e0787d9d4402c7952d8 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf7758708864218b81588bec3b25cc2 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40d4d9936e14ecd8332d6b897af4e3a : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6115c7ecd22f4a559f4955321570d356 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acfd479652c04d27a1cbe5939f141b58 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f79142f2009469a9ba12535cb36d164 : SUCCESS in 7m 15s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7630841a9fba47e996f13afa8e281460 : SUCCESS in 46m 51s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3d81f376cb594e158eea7db6a18a429e : SUCCESS in 58m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dad14d2589a647e9a4b5474f7b3c1c56 : FAILURE in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f551e43a364c1b94966026415b2ab15398bb218c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 21:14:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81c2f8b4c2234368da8015fdfa3c37e8f2a64ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 21:14:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0225b172b69e288f6eb19e903112f55fc9ca33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 22:00:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee5d9232e7c74d38be3b7447f158a1e2 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4be297200be45ae8fa6200ff2c43dd5 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac9b3dfd7ebf4243996fdcd43d8d6329 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2569a7a49744ee38aaefffffd778222 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b2167c0a3b47379ca7d5e726a73359 : SUCCESS in 9m 00s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/04e7f49e6fea4ad0ba96e63916630f5a : SUCCESS in 40m 31s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/089259a9de9d45d8bf38c1788fa246e8 : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a6557ff7fccf4d7d45e5a043b8bf4cba62b5e2f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 22:00:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6ef16f95afbdb184a647baecd24406ab9e788932","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 22:06:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a404f59d3a4217b677ca52c87e8e2e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37bee67efde42f9e3926e74b0f9e29f94f843c62","revisions":{"81c55e5fbff4ef307d1ce6313731d451312597f2":{"kind":"REWORK","_number":1,"created":"2021-01-20 09:00:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/771576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/76/771576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/76/771576/1"}}},"commit":{"parents":[{"commit":"0913a41c43be3e2b647dda4a19544e539f8cf801","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0913a41c43be3e2b647dda4a19544e539f8cf801"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 08:59:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 08:59:52.000000000","tz":60},"subject":"Make lower-constraints job non-voting","message":"Make lower-constraints job non-voting\n\nIt is preventing us from merging some patches. Disable until it can be\nfixed.\n\nChange-Id: I55fbae296f5582148f5b48b7cf96fbd77d014227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/81c55e5fbff4ef307d1ce6313731d451312597f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/81c55e5fbff4ef307d1ce6313731d451312597f2"}]},"branch":"refs/heads/master"},"37bee67efde42f9e3926e74b0f9e29f94f843c62":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-20 20:06:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/771576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/76/771576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/76/771576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/76/771576/2"}}},"commit":{"parents":[{"commit":"b1841dc4b90d9127884379d4fe1cccf99c7a579c","subject":"Fix use of legacy Alembic API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b1841dc4b90d9127884379d4fe1cccf99c7a579c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 08:59:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 20:06:38.000000000","tz":0},"subject":"Make lower-constraints job non-voting","message":"Make lower-constraints job non-voting\n\nIt is preventing us from merging some patches. Disable until it can be\nfixed.\n\nChange-Id: I55fbae296f5582148f5b48b7cf96fbd77d014227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/37bee67efde42f9e3926e74b0f9e29f94f843c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/37bee67efde42f9e3926e74b0f9e29f94f843c62"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
