)]}'
{"id":"openstack%2Fblazar~778485","triplet_id":"openstack%2Fblazar~master~I464f5388646de0220b9309d5200060fe5a146eb1","project":"openstack/blazar","branch":"master","topic":"revert-reuse-rpc","hashtags":[],"change_id":"I464f5388646de0220b9309d5200060fe5a146eb1","subject":"Revert \"Re-use rpc transport\"","status":"ABANDONED","created":"2021-03-03 15:16:12.000000000","updated":"2021-03-12 10:55:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e92a62cbd1c524e0894a6aa095b9c05538a5647","_number":778485,"virtual_id_number":778485,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-03 16:08:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2021-03-09 20:39:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 16:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a2bec8aab1fef5cdfbce426ba46eacdf88dd78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-03 15:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"869a88783c25bcd1ebccdcacd883be36b3eb6b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:08:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff0b3281615c478ba6d013a1b3370e9e : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7186652aee4647119bcacb9ee7210ee7 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c6fe0cda1142f697fe34ce18872b3c : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a5a4ac32b4f45298d7a793c832bea03 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3609f7cd183047de86daadc885b96afd : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d351bf456de84e6085e61eba38392c33 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a794d930bf354100bad62f05941c93d0 : SUCCESS in 14m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6acdfdc2512140d9a7ba11af3222c7a2 : SUCCESS in 38m 32s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ad4a8c113439433c9cf63279c6f6e8b9 : SUCCESS in 47m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3420bbd574b14e99baae109285787b30 : FAILURE in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eef096d36708fcdb3adef8562723568fc50c952e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-09 20:39:55.000000000","message":"Patch Set 1: Code-Review-2\n\nThis change doesn\u0027t seem to help. The issue probably originates from oslo.messaging and would be fixed by Id5cddbefbe24ef100f1cc522f44430df77d217cb. I will abandon this change once confirmed.","accounts_in_message":[],"_revision_number":1},{"id":"58fd5c82f1f9a98c3b384797e3b3b776f513c705","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 10:55:46.000000000","message":"Patch Set 1:\n\nConfirmed, oslo.messaging 12.7.1 fixes this issue.","accounts_in_message":[],"_revision_number":1},{"id":"8e92a62cbd1c524e0894a6aa095b9c05538a5647","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 10:55:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b928df238c096bf6bb18d2b41ddf175ebe2e1d","revisions":{"18b928df238c096bf6bb18d2b41ddf175ebe2e1d":{"kind":"REWORK","_number":1,"created":"2021-03-03 15:16:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/778485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/85/778485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/778485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/778485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/85/778485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/85/778485/1"}}},"commit":{"parents":[{"commit":"2d95b84e7db3c26bfd08efe66ea1dcc7654670f3","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2d95b84e7db3c26bfd08efe66ea1dcc7654670f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-03 15:15:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-03 15:15:39.000000000","tz":60},"subject":"Revert \"Re-use rpc transport\"","message":"Revert \"Re-use rpc transport\"\n\nThis reverts commit 7d131c96d11799a463e8e8e0d15251175fb48ea1. We are\nseeing oslo_messaging.exceptions.MessageUndeliverable errors after\nseveral hours of runtime in DevStack.\n\nChange-Id: I464f5388646de0220b9309d5200060fe5a146eb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/18b928df238c096bf6bb18d2b41ddf175ebe2e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/18b928df238c096bf6bb18d2b41ddf175ebe2e1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
