)]}'
{"id":"openstack%2Fblazar~805461","triplet_id":"openstack%2Fblazar~master~I464898bdeda6b55a028c8a14a54eb137c92a83c1","project":"openstack/blazar","branch":"master","topic":"change-805461","attention_set":{},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2022-01-27 16:16:29.000000000","reason":"Jason Anderson replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-27 17:04:02.000000000","reason":"Change was submitted"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2022-01-27 17:04:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I464898bdeda6b55a028c8a14a54eb137c92a83c1","subject":"Update list_allocations in hosts plugin to expose start and end dates","status":"MERGED","created":"2021-08-20 21:35:40.000000000","updated":"2022-01-27 17:19:55.000000000","submitted":"2022-01-27 17:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805461-change-805461","meta_rev_id":"97b0972a2295309ad106f7e1ea11e2dd5d120b0f","_number":805461,"virtual_id_number":805461,"owner":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 17:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 16:16:29.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2022-01-27 14:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 16:16:29.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":1,"date":"2022-01-27 14:41:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 22:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 17:57:54.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2021-12-16 16:19:16.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c9e743463fb37db2974bc19ba00c54dc2145c402","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-08-20 21:35:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae60145044a31e38213500370775521743ee386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 22:18:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae84d960630452ba8efeef0e3a7531f : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da61f5086f664d93bbe56f02d1692947 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ab7d45fd0e405688b367ba49b6e6eb : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76250dede38044209602c76cac9bfa44 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdfa28aacf02407cbfe3226b95002c89 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a46a87a8fef941d89ec3073d0bb70ff8 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c716879a897c48469f7c4307e509d911 : SUCCESS in 8m 40s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/04dc7c09a23a404e943813fe0dfdb785 : FAILURE in 38m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/2d0b1440bf444e4db9d658564cd2ccbb : FAILURE in 39m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6283b4fe637546a1ad7318e025925eab : FAILURE in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69e1616a027b4c791a2a01f4332ca2a238623539","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-08-31 16:39:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4bd13e87b66ec72e26a941c40c99432a5be11c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 17:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132603c2eee94d58a0c3ffad0129424a : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c68a6570c5cb474ebc529ebda4f214da : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d42523300548509e696405b02ff896 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1f72a59fc34f9186a95bc50206d916 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f9553e8711c4d4288de9e51c614cbde : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f74897d0114eeead0614d52f94187f : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7e3f82258f4dee91fc71fda69702cd : SUCCESS in 11m 25s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/dd7606d9162545d9a3bbcb554453755e : SUCCESS in 38m 18s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/fd519816f9e44065bf58dc92bf798c38 : SUCCESS in 45m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a655efe5b95149a7bb43a99b8d94ceb7 : FAILURE in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"759e26ace9fa421330edb3713b714e35f9ec3a48","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-08-31 17:37:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"168f1aab5ac153680befac5cf8fe2ebe8e05f9e1","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-09-08 17:57:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nThose tests were a bit hard to verify by reading it, but they seem correctly updated.\nBlazar does support API microversions. I think this should probably be a new microversion. I am not sure precisely how to do microversions in Blazar, as I haven\u0027t myself, but I think there are some reference commits and also some new patches that introduce microversions.\n\nThanks! See notes about the query logic","accounts_in_message":[],"_revision_number":2},{"id":"2b576055000017edf9568ab52421f5b94714f8d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-15 22:04:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bdfb8c3eca7a7ee7de3df0f570ce3973b10f1ea2","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-15 22:05:28.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d659da0de5921f661c6ad5c1740c72fabd9f4c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 22:57:55.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c022a032e8469693f3a856ee9a40e4 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f743d5a3637448eeba7aa02092d5e4f3 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccafc97a31741d99335d6c1b8434575 : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8837fd4886f4befac0d2d768a9e460e : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ea295ea51924e14815d0ab483d0baa6 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ba3fa5c219742cd82557f8c31511da0 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed46c837d214d5283dcd7ce02ac5c39 : SUCCESS in 7m 29s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0de7a43a258d47898487c62cb0c2c4b8 : SUCCESS in 44m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/41bb2689756b4a55a31e74d81f26010a : SUCCESS in 50m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58c32b86cfde4047b2d72b06eac6977a : FAILURE in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f37a014c3a459f6aa104699863ae994a49cf7873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-16 15:54:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff74d873b223f5c54b357e8d443e0b423bb30e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 16:50:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8744996b4349474a9ec3782f1c72f6d9 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/218641d7718b4ed3b2abc3f304f60c2b : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c11abe8bbcc4d37b7e266824b28579a : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d19ff003b9b743ca87b03d53ab34f50a : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bc597b0a57f4467aa933afd4cb54072 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d84e3c401b9b4895ad8433c7c3a83aac : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a637111363ce48c6b029dcd71a1357cc : SUCCESS in 8m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2d81ee93bdc34f9888921875e249c37e : SUCCESS in 48m 55s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/501943053d4a4266999ca0e6f93fc79f : SUCCESS in 36m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d60341a8f3554ca2a603382cba430ec4 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be70aa4a45017748892e82031c0692d2f6e9208d","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-12-13 20:00:42.000000000","message":"Patch Set 4: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df05d27d250cff6c13ab4055ecf993f9733f0176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-16 17:46:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6f69040ef5d5c847aff51d2715a757bdc8e0963c","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-16 17:47:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"674e1e17c64afda8124eafb1fb1aee5fd7abb402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:44:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64f88be7b3c744f79156cebaaca6bd20 : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/679295f3b52b4938b77fcdca20855483 : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c301c29402744897bfc89bd4c9545238 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3808787d8b7549aba0706cdc943a558b : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd3a2a28f184218b1b45dac5a857865 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b486ae2a0df44e4b6e87a5b44024cb9 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c5f09c5d8f4c5893d8646fce5c652e : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b4fd6f7d792493cb514be28cffa6438 : SUCCESS in 5m 33s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0d52220520044384a800571378598ff2 : SUCCESS in 47m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b00bb104c369489095e41386fc605da1 : SUCCESS in 39m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dba2facd2d6842a3a14d20328d9b9188 : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"63d750d863971f46f5dc1298951b5de57a5039e6","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-12-20 23:35:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d895583f1e511ea82499d0c4dfcae5176810da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 14:07:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8cd593651e2eb27a860cf1adbb482eeea97ed54","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 14:41:27.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac01cf116382980e05198a2fa9dea921e8e0aaf0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 14:41:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"19de43c639c1c0e5098c93b22b02aceecb505545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 14:58:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26e9105ec7bb4132aa105476d6fbf821 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c218f692b24eeba43262f359971f8a : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/177ea354bb344f20a9e9de0053761157 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb0842245f64981b6d69a42ccc30259 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05bb9293f5f425189adcaa31dce7901 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38fa2fa4d046400eaea868a568b46a3e : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097e1477cc4d49f3ac5d68e5712fb696 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/199ddd3746714181a82b96b7a756ddbe : SUCCESS in 6m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ddedf0eb7a40475cac3792c3d7ae3e85 : SUCCESS in 43m 03s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec846483f774778a57239f1cc0a5e2b : SUCCESS in 41m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af2c9bef33224b55ac0e25342a271bd6 : FAILURE in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5b8617a818ae54022c508286d99bb4fe79c9092b","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-01-27 16:16:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dcfae1989365741cd36c0f01a46c35070a054a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:24:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"262d4849f0101c167190e7c26aacfab91c7207ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:04:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b81b192a34884b75a13bff6614738c07 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78c224aa41ec452e85d63c2fbe4704f7 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36931121a8784e07ac9432b5c3e78fd9 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afcc08248d6744e8a6ac2c8a830bebb7 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67afccd392f54ed9a65b205dcc486512 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8d8976d80a4f689230dbd6267afa24 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568f83ce2b184268a9878394bd5893df : SUCCESS in 5m 20s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0501d33aa6da4988b13d6a721fca9adc : SUCCESS in 32m 11s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0f5d6b1350ff41509fff5174e0f7ed8d : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":6},{"id":"983d2f2934702ef902cd5da068682e47d69294c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:04:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"97b0972a2295309ad106f7e1ea11e2dd5d120b0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:19:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0ba952b8df024e5fa5cda784308189d9 : SUCCESS in 1m 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c81ca4777648b2a04d9f051b49bf30 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d35ea4fe516456889a217c8e4c6008e : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bf38e16ec3807536a421ba9d6249171742f1b960","revisions":{"03db0e17cd148e0cac4b55f733e104202a45d141":{"kind":"REWORK","_number":1,"created":"2021-08-20 21:35:40.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/61/805461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/1"}}},"commit":{"parents":[{"commit":"233e1c146d77cf0b2944f9dd24fa5d25423dce79","subject":"Merge \"Handle AggregateNotFound when deleting aggregate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/233e1c146d77cf0b2944f9dd24fa5d25423dce79"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:35:39.000000000","tz":0},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows the host plugin to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view)\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/03db0e17cd148e0cac4b55f733e104202a45d141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/03db0e17cd148e0cac4b55f733e104202a45d141"}]},"branch":"refs/heads/master"},"cdffdb9300d89c4c3854cdf1987149676e318c27":{"kind":"REWORK","_number":2,"created":"2021-08-31 16:39:39.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/61/805461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/2"}}},"commit":{"parents":[{"commit":"233e1c146d77cf0b2944f9dd24fa5d25423dce79","subject":"Merge \"Handle AggregateNotFound when deleting aggregate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/233e1c146d77cf0b2944f9dd24fa5d25423dce79"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-31 16:39:37.000000000","tz":-300},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows plugins to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view)\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cdffdb9300d89c4c3854cdf1987149676e318c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cdffdb9300d89c4c3854cdf1987149676e318c27"}]},"branch":"refs/heads/master"},"dcebec9e85dd86d9e7c9a6865c62c216ea750e48":{"kind":"REWORK","_number":3,"created":"2021-09-15 22:04:35.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/61/805461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/3"}}},"commit":{"parents":[{"commit":"233e1c146d77cf0b2944f9dd24fa5d25423dce79","subject":"Merge \"Handle AggregateNotFound when deleting aggregate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/233e1c146d77cf0b2944f9dd24fa5d25423dce79"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-15 22:04:26.000000000","tz":0},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows the host plugin to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view)\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dcebec9e85dd86d9e7c9a6865c62c216ea750e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dcebec9e85dd86d9e7c9a6865c62c216ea750e48"}]},"branch":"refs/heads/master"},"8b647c4312b6d2c9c57a3427d367797c70e02d33":{"kind":"REWORK","_number":4,"created":"2021-09-16 15:54:04.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/61/805461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/4"}}},"commit":{"parents":[{"commit":"233e1c146d77cf0b2944f9dd24fa5d25423dce79","subject":"Merge \"Handle AggregateNotFound when deleting aggregate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/233e1c146d77cf0b2944f9dd24fa5d25423dce79"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-16 15:53:52.000000000","tz":0},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows the host plugin to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view)\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8b647c4312b6d2c9c57a3427d367797c70e02d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8b647c4312b6d2c9c57a3427d367797c70e02d33"}]},"branch":"refs/heads/master"},"58e05c30af2f380ce8efa17fa732011ce889c034":{"kind":"REWORK","_number":5,"created":"2021-12-16 17:46:35.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/61/805461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/5"}}},"commit":{"parents":[{"commit":"233e1c146d77cf0b2944f9dd24fa5d25423dce79","subject":"Merge \"Handle AggregateNotFound when deleting aggregate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/233e1c146d77cf0b2944f9dd24fa5d25423dce79"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-16 17:46:34.000000000","tz":-360},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows the host plugin to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view)\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/58e05c30af2f380ce8efa17fa732011ce889c034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/58e05c30af2f380ce8efa17fa732011ce889c034"}]},"branch":"refs/heads/master"},"bf38e16ec3807536a421ba9d6249171742f1b960":{"kind":"REWORK","_number":6,"created":"2022-01-27 14:07:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/805461/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/61/805461/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/61/805461/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/61/805461/6"}}},"commit":{"parents":[{"commit":"39c67f09b8095c4d8aadfe550f0aa924a3693f5a","subject":"Fix list_allocations for multi-host reservations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/39c67f09b8095c4d8aadfe550f0aa924a3693f5a"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-08-20 21:19:05.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-27 14:03:24.000000000","tz":60},"subject":"Update list_allocations in hosts plugin to expose start and end dates","message":"Update list_allocations in hosts plugin to expose start and end dates\n\nThis change modifies the DB utility to fetch reservation allocations so\nthat a list of reservations is returned with host_ids as an attribute to\neach reservation.\n\nThis allows the host plugin to include more information when listing\nallocations, namely start and end dates. This can be used to see when\nresources are available, and for the creation of a reservation calendar\n(blueprint calendar-view).\n\nChange-Id: I464898bdeda6b55a028c8a14a54eb137c92a83c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bf38e16ec3807536a421ba9d6249171742f1b960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bf38e16ec3807536a421ba9d6249171742f1b960"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
