)]}'
{"id":"openstack%2Fblazar~810298","triplet_id":"openstack%2Fblazar~master~Ib9f1140c44c5e4fbef6e019c48a842869368cb21","project":"openstack/blazar","branch":"master","topic":"bp/resource-properties-discovery-api","attention_set":{},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2022-01-04 21:02:41.000000000","reason":"Mark Powers replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-21 22:12:00.000000000","reason":"Pierre Riteau replied on the change"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2022-02-21 22:54:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9f1140c44c5e4fbef6e019c48a842869368cb21","subject":"Add resource properties discovery API","status":"MERGED","created":"2021-09-21 17:33:04.000000000","updated":"2022-02-21 22:57:43.000000000","submitted":"2022-02-21 22:54:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"810298-bp/resource-properties-discovery-api","meta_rev_id":"9e03f5be2280b56d04ab3b1cd14f71538bb2664a","_number":810298,"virtual_id_number":810298,"owner":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 22:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 20:43:02.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2022-02-21 22:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":1,"date":"2022-02-21 22:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 18:41:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 21:48:02.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-12-15 21:48:02.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2022-01-04 21:02:41.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REMOVED"},{"updated":"2022-01-04 21:02:41.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REMOVED"},{"updated":"2022-01-13 14:52:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-15 20:43:02.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"851c4e6b5c25f9d25f6653e86b4e059fce9de684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-21 17:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e237bfa7945d2fa4c52df0ef068779ce75c678","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-21 17:45:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"cda8716be264a54685a747dce8f303004d3fb4dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 18:41:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e4265e75503474b829041ea89f2a60f : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8c62d8601c74467b8189c35e4437506 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf3ae96cf6943548ba7b212e12d7e73 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d46bbbc7a9046b29615a4929e75476e : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e87c3d0a663443b3823c261cc0353644 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/050d8851b3884416a16db55cba1b7720 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5971fbfee28b4521abf515fa09085afd : SUCCESS in 8m 00s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/207bc3c6d3c24204855a4e10b366af4c : SUCCESS in 1h 03m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc5b7712e8742588de65a2360db897a : SUCCESS in 40m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0de134eed8324c998f8fafabcfe00830 : FAILURE in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2eefefb20b6517f7d30dfe188428d2600bb40a57","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-09-21 21:23:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"9b4c1003de82345d7be03c757acc6a6759959d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 07:57:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"62e7a6ac64f235521ac24dfc3a690c1946b0318f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 08:34:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65cf86a6890042f19bf5b2b28565fd83 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4b6407f30e14451929b8cb817605e5e : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d182e3bfd1b040fcaacead6532c57336 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a50734f177d949d0812304c8ade0cc82 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/033c03542ff449ed87998f1155b9c734 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d944f75816d47a7ae8d7b9d59705b8f : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806edcd7027a4a8db5b6cfebbc731e06 : SUCCESS in 8m 00s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/3bec912872044c039f84460e460e5e15 : FAILURE in 31m 18s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee9c8dbec294cda8dd2d13307b540a6 : FAILURE in 28m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7dbc601e3546481c97588cbb22803bc7 : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d1bf73e1e75844f4f855655373ef3f289a02cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-02 19:47:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3e644b22d25a20240b319f2b50af15feb3d7a1e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:39:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448fec577881426d96e6bfaf50692154 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6c5d66a4ee54886a787761e642a5e27 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6053d55ef7264007961105c11c04570e : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d72ea92970a4f6888de23a0499ae1d4 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/698ea9d059354efcb27e599aa5e1bd44 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e964c5d81a4ff79afea98797d0349f : SUCCESS in 7m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f022e5abea524d229648c5f88abdcebb : SUCCESS in 43m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6c4b7d74ffee4041ab01e514f77f1ea6 : SUCCESS in 37m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e9d30197bc1c4492903b5abbe2636493 : FAILURE in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65c20dafe81af7d627d36613aa633dd1e8bef4e0","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-12-20 23:50:10.000000000","message":"Patch Set 3: Review-Priority+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51d8a63f0492ec6a52ccb9e0fecc0583dd13daa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-21 23:03:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0dd2a6d7e18111b7b6fbc024bdcd75432b1af5e8","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-21 23:03:22.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a0efccc27c7572a542d1ed92d7d8624f03faf7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:52:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d8e580d0818410a8443ca34c899deaf : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5feeeaded95d48c9a181603ec89b36b3 : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c849e5a52f5a4088a7027ab9a3d13b47 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a6055ec66d44ef9d05f8757863673c : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea5e7b2539504b3c829ad1e0a6b486ad : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50807c69857c4188a902840c1b7ac423 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c557ab0d465491298fb58588007cc81 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdce42c632794d7e891cd3b9fe7b3afc : SUCCESS in 4m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ce0895b221b048a69f7b70ebc8e2c5b1 : SUCCESS in 43m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c0e801e5b90e470598a2ad76787dae8f : SUCCESS in 43m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88595aa6c9c54108bfd20c8be6e955d9 : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00f176e26a14403c0c1537d111673b6a2f990cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 17:58:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"24531b5dbc2d9ce0a5d0d64921ea8199b95f7ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 18:08:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"292313dd86c2fe2c7c2daccbfe5859bea4053786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 18:08:35.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cfdc64166064b6c47b6fa1bd5836fe0829393764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 18:47:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9641d8cd1ff0439d91b77bb13170bcd5 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2353aa3e226a4153a34d26a224be4154 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e08b17921a48db8be7779209073e7e : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87decb6f04894901a15545428192fda4 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0a126af38b745b4b1005cf33f6572d9 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74a9d610c4d4416998e4cfa3b4bc70b : SUCCESS in 7m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c1313b4ce69f40518a18adcf8a210715 : SUCCESS in 37m 21s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/bbc99acd52044b98a274c1661397914a : FAILURE in 25m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2c2511e0c2044f883043b47479ff6b7 : FAILURE in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"36df69f90f0732e718161cc6885c541a38188888","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 21:02:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a854337aadc079c4f06d4c618e356a7e700853aa","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 21:02:41.000000000","message":"Removed reviewer Pierre Riteau.","accounts_in_message":[],"_revision_number":7},{"id":"1b2141225571747fd71468d75266a215fd1ca803","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 21:02:41.000000000","message":"Removed reviewer Jason Anderson.","accounts_in_message":[],"_revision_number":7},{"id":"4bc65ae0dce0f6c980e65e695ef9607b23f1edd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:51:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc4f29af2957402eaa70baabdfa05f00 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bb012b981d448c2ba631ead6cd85bf6 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20376c1b8274fe5933601740bc89280 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/346d26c415784da08df28462dc91ddbc : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/608d54a728cf4caba3151dcda5fc9107 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac3894e71c54438ba1ebc572ebe279d : SUCCESS in 7m 03s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/016e8397e56d4953aed9f450522c9fab : SUCCESS in 43m 11s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/04af2ee8665d460a8a8e14e0169cb9db : SUCCESS in 46m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ef36aea1f614498bc3435d25ee6d313 : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"22e7296c5b93912b8ae9f38932e09eb22f5bd783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-11 22:34:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"074a3f91995968ad72cd2ea56a138d61d135fa5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 23:30:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda5d3bc217e429e869de5c760857e69 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea783721235a4ad08db19ce8697a074b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4f5da34cfd4529a7369cb69b5ede3d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05989b31a87d4dec87a4040eea9594c3 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/419c41de0c1846c7806c6e67826e9fc0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521338600ac942bebb28c831cfd8952b : SUCCESS in 9m 14s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/51427b8e8c704872b994c5e30faacea1 : SUCCESS in 43m 26s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/71fe3325bf55414eb83ff588d3355d94 : SUCCESS in 48m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdf622c8941446ff9047156da5216c49 : FAILURE in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ae0b0776bc259d569be27e8d767888abdea8b618","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 14:52:39.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"07d379ea313246cd680e85d53807521068729ef2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 15:28:37.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"717dcd2b22601b5f3c723d4f989c52038a9094bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-14 16:23:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bda96c0bf09a57380d8d2f073949332c8ad8d0b9","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-14 16:24:20.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c4cd1696933c700745988728da5f1a47f1976454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 17:19:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/55188bae06714724b726a7768624b098 : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95ccc482e0184e7c8c3251ea759aa60a : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b674a1914a54be6b3f92ab755591ec6 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e348240ee5a44a03b47879e5f57ad082 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a647abc4986e442e8ddf3e85587b870f : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/060db511b7b947efa8cc54a7397a6b29 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10d9c9a5bf64261974bef87a88a88bb : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/268a5d11e4cf400dad7e85f4989fc5bf : SUCCESS in 5m 17s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d9ed2799fbd74c97ab94cfd3e1cc4daf : SUCCESS in 43m 16s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a5ca96a35db4429db9d4747d012b0143 : SUCCESS in 47m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ab0cbb60e2a4b3e913c72289e0291d0 : FAILURE in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4316829b40ae2e3fef6367a60a40668cdbfa724c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-25 17:51:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"70d46825a14f1395bce532d753e676f759fe937d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:37:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0aff64327b74520a6d88c934f8b43a9 : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/789968397ad6487b8f703d0a81eea55d : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d959be771960429f80114586037cfc35 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/720a1cd2f2974033ac6fcb773e493725 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e227976afdbe4843aa1212fee3969fa4 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb971e10b9e4c529b36133c4756fb4c : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f72eb3701d4463eba8efdad5e484259 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac7c440a271470d9daf8e7d72070524 : SUCCESS in 5m 13s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cdfaecdff2354dff9448ab84c11a4924 : SUCCESS in 39m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/485d92d8e6ff4c47bb3dd9e6ddb891b0 : SUCCESS in 40m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d14ba85dc0c41739e544ee369cdf2d0 : FAILURE in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3eb172e52e80b55ee43a37794e6ae8811cad4e48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-28 17:33:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5ccdddd00463f9122d624535e06955a901a5c523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 18:58:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebd57d531ebd4626a8a4e50c8acd8f7c : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d3574307a4e490e8c159e55563809f5 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7209524835a640f1871e3a21ba25cd0c : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c865817b449417fb3832d7a0af2740b : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dde4f2d97a442f1a79d9bac73b99b56 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0400230048294288aa7a6fbaa798a9b0 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22dfba794adb4de3b27fafdd619e56ce : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9807cb5daaa043238630c255c246cacf : SUCCESS in 7m 23s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8d623db9f19e4419bf175f65ec1d401e : SUCCESS in 45m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f9365e0a83f0465cbfa08b346d9692ed : SUCCESS in 48m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95dd5a1ba7594e1190cc1498fa5320fd : FAILURE in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"18597ce99ed119c5e3e9044cae1d986d077f367a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 16:22:30.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"aab3ca12bb6b8e5ef8f807e88cf33a2ad2d5ba2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-02-10 17:23:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8541f764dcd1d358c383c64c99d33bab360919ca","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-02-10 17:24:43.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"26de5f4b4ba143eff57e2e810256edba670ec0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:26:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/118434b75b6b42108eaf79140b802f81 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b68651bd1b3744ddbcb32fb289dca1f9 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c99ee7e11294c7f9d730467c4f59391 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa26a7097c1a47589a5d1f8de9b7113d : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac084ad43f9469aa57a733eb3e92372 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42c0245acfdd429abf346c4ed50f5517 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9c74273a9d42b984510f7180aaf904 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79654731cf9a4eb2925f08650c079d13 : SUCCESS in 4m 33s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c9c9483686714ed9bff0f8f3d21c03c0 : SUCCESS in 31m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/09b2a08521dc43c9996b60cd3f59922f : SUCCESS in 42m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14b84a876dd9484fa3e9e48e35df687d : FAILURE in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8e491f1016062cfe7339b81b13de33b359b95a79","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-02-15 20:43:02.000000000","message":"Patch Set 12: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d0d0977fffd8740d8ec82ccdf0a5d62830d60a03","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:12:00.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"49b9bf9c4f4f71be12bf2128252f230d50d23fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:12:47.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"90d60344c65a6ff25953fe919fb207ecd5ae5fe3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:13:20.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"76793197aa61e341fd740c06ec82631e7142cfea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:54:51.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a1bc24bcb2c4443bbf5cd2f40b78c5fc : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fee341de2f4843a0b60b8a13e70dba0a : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e900a1ce1e4613975a93998ca9af17 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a04f7b8d5ca143efb82d0de02167b16c : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20dfcf6653f1412587c7725930f6aed1 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e76dfcf4a5472b83f995783ff2bb69 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05222da19c5c44feaa5c70cc903f5108 : SUCCESS in 4m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/73e9c289460e4915b9787a071445f2f0 : SUCCESS in 35m 40s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b7e5b83caa5242b0b4b89cfc35d57cc7 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":12},{"id":"5142c8363ea3d8d28bdd5d496d02c7474c9d7680","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:54:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"9e03f5be2280b56d04ab3b1cd14f71538bb2664a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:57:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/da0f09ec76704e1cb341bbe5e87abaf6 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a91e51336cc4315b086fa0d59eaab08 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50c4ad16c89e423dbf724baf0e9141fd : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"f5e6d2482681e2519dc831254ac34352435cbafc","revisions":{"80f61d4ad9779d34991da15eb37e62e9a937d834":{"kind":"REWORK","_number":1,"created":"2021-09-21 17:33:04.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/1"}}},"commit":{"parents":[{"commit":"370f2afb8862052c59ad3a0c7001923d1e8fdf30","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/370f2afb8862052c59ad3a0c7001923d1e8fdf30"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/80f61d4ad9779d34991da15eb37e62e9a937d834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/80f61d4ad9779d34991da15eb37e62e9a937d834"}]},"branch":"refs/heads/master"},"f3e1d0142af4dd78b31927f940b924d03f9f5adc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-04 07:57:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/810298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/2"}}},"commit":{"parents":[{"commit":"949c52dba434d6888b14c35c4db687118f0c74e0","subject":"Ignore usage enforcement release note for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/949c52dba434d6888b14c35c4db687118f0c74e0"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 07:57:01.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f3e1d0142af4dd78b31927f940b924d03f9f5adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f3e1d0142af4dd78b31927f940b924d03f9f5adc"}]},"branch":"refs/heads/master","description":"Rebase"},"7288e66c2cd0d4969d14d335c6eaa12076be0c63":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-12-02 19:47:43.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/3"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 18:39:01.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7288e66c2cd0d4969d14d335c6eaa12076be0c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7288e66c2cd0d4969d14d335c6eaa12076be0c63"}]},"branch":"refs/heads/master"},"f0a4b1051609fdf64a92436bac36e21a4e64a280":{"kind":"REWORK","_number":4,"created":"2021-12-21 23:03:03.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/4"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-21 23:01:01.000000000","tz":-360},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0a4b1051609fdf64a92436bac36e21a4e64a280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0a4b1051609fdf64a92436bac36e21a4e64a280"}]},"branch":"refs/heads/master"},"7fa7be28c6067a18ed593c5749f979839f4f695a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-04 17:58:26.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/5"}}},"commit":{"parents":[{"commit":"d85abc7589482dea6bc202606a98226776fde54a","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-04 17:58:26.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7fa7be28c6067a18ed593c5749f979839f4f695a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7fa7be28c6067a18ed593c5749f979839f4f695a"}]},"branch":"refs/heads/master","description":"Rebase"},"03e6b78f32371e23b1ff87f8bc6d97741c5e3f3f":{"kind":"REWORK","_number":6,"created":"2022-01-04 18:08:07.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/6"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-04 18:07:59.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/03e6b78f32371e23b1ff87f8bc6d97741c5e3f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/03e6b78f32371e23b1ff87f8bc6d97741c5e3f3f"}]},"branch":"refs/heads/master"},"a55e954bd55e53d3df4d839d99c12f166faa71d9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-01-04 18:08:35.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/7"}}},"commit":{"parents":[{"commit":"d85abc7589482dea6bc202606a98226776fde54a","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-04 18:08:35.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a55e954bd55e53d3df4d839d99c12f166faa71d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a55e954bd55e53d3df4d839d99c12f166faa71d9"}]},"branch":"refs/heads/master","description":"Rebase"},"10072b1e68327f8bb03a6064d2ea2fa91fa4ace1":{"kind":"REWORK","_number":8,"created":"2022-01-11 22:34:40.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/8"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-08 21:42:45.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/10072b1e68327f8bb03a6064d2ea2fa91fa4ace1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/10072b1e68327f8bb03a6064d2ea2fa91fa4ace1"}]},"branch":"refs/heads/master"},"20ba175d1c2ce3f93f905d11cd116c8c3d07be35":{"kind":"REWORK","_number":9,"created":"2022-01-14 16:23:00.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/9"}}},"commit":{"parents":[{"commit":"d85abc7589482dea6bc202606a98226776fde54a","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-14 16:22:31.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively. Two new\nAPI endpoints are added for hosts, ``/properties`` and\n``/properties/\u003cproperty_name\u003e``, which allow for listing available\nproperties, and updating a property respectively. Properties can be\nlisted with detail, showing possible values and visibility. Admins can\nlist public and private properties.\n\nA new database table is added ``resource_properties``, which stores\nproperty names and resource types. Resource specific property tables\n(e.g. ``computehost_extra_capabilities``) entries store a foreign key to\n``resource_properties``, rather than the capability name.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/20ba175d1c2ce3f93f905d11cd116c8c3d07be35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/20ba175d1c2ce3f93f905d11cd116c8c3d07be35"}]},"branch":"refs/heads/master"},"21c020ea2ed022e999aa5d3cbfeaef5210162c2e":{"kind":"REWORK","_number":10,"created":"2022-01-25 17:51:32.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/10"}}},"commit":{"parents":[{"commit":"39c67f09b8095c4d8aadfe550f0aa924a3693f5a","subject":"Fix list_allocations for multi-host reservations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/39c67f09b8095c4d8aadfe550f0aa924a3693f5a"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-25 16:13:42.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively. Two new\nAPI endpoints are added for hosts, ``/properties`` and\n``/properties/\u003cproperty_name\u003e``, which allow for listing available\nproperties, and updating a property respectively. Properties can be\nlisted with detail, showing possible values and visibility. Admins can\nlist public and private properties.\n\nA new database table is added ``resource_properties``, which stores\nproperty names and resource types. Resource specific property tables\n(e.g. ``computehost_extra_capabilities``) entries store a foreign key to\n``resource_properties``, rather than the capability name.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/21c020ea2ed022e999aa5d3cbfeaef5210162c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/21c020ea2ed022e999aa5d3cbfeaef5210162c2e"}]},"branch":"refs/heads/master"},"75ec9d7072e550bb4965fb619f60ccb0b54c1fd4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-28 17:33:31.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/11"}}},"commit":{"parents":[{"commit":"4a2bb67fbf5eec4ccf568b4c3d6372c92171143b","subject":"Merge \"Update list_allocations in hosts plugin to expose start and end dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4a2bb67fbf5eec4ccf568b4c3d6372c92171143b"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-28 16:03:45.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively. Two new\nAPI endpoints are added for hosts, ``/properties`` and\n``/properties/\u003cproperty_name\u003e``, which allow for listing available\nproperties, and updating a property respectively. Properties can be\nlisted with detail, showing possible values and visibility. Admins can\nlist public and private properties.\n\nA new database table is added ``resource_properties``, which stores\nproperty names and resource types. Resource specific property tables\n(e.g. ``computehost_extra_capabilities``) entries store a foreign key to\n``resource_properties``, rather than the capability name.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75ec9d7072e550bb4965fb619f60ccb0b54c1fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75ec9d7072e550bb4965fb619f60ccb0b54c1fd4"}]},"branch":"refs/heads/master"},"f5e6d2482681e2519dc831254ac34352435cbafc":{"kind":"REWORK","_number":12,"created":"2022-02-10 17:23:38.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/98/810298/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/98/810298/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/98/810298/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/98/810298/12"}}},"commit":{"parents":[{"commit":"4a2bb67fbf5eec4ccf568b4c3d6372c92171143b","subject":"Merge \"Update list_allocations in hosts plugin to expose start and end dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4a2bb67fbf5eec4ccf568b4c3d6372c92171143b"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-09-21 17:31:43.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-02-10 17:23:15.000000000","tz":0},"subject":"Add resource properties discovery API","message":"Add resource properties discovery API\n\nThis allows users to query for resource properties of hosts,\nallowing them to be used in creating leases more effectively. Two new\nAPI endpoints are added for hosts, ``/properties`` and\n``/properties/\u003cproperty_name\u003e``, which allow for listing available\nproperties, and updating a property respectively. Properties can be\nlisted with detail, showing possible values and visibility. Admins can\nlist public and private properties.\n\nA new database table is added ``resource_properties``, which stores\nproperty names and resource types. Resource specific property tables\n(e.g. ``computehost_extra_capabilities``) entries store a foreign key to\n``resource_properties``, rather than the capability name.\n\nImplements blueprint resource-properties-discovery-api\n\nChange-Id: Ib9f1140c44c5e4fbef6e019c48a842869368cb21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f5e6d2482681e2519dc831254ac34352435cbafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f5e6d2482681e2519dc831254ac34352435cbafc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
