)]}'
{"id":"openstack%2Fblazar~816627","triplet_id":"openstack%2Fblazar~master~I5fdd5e6095e579e2584984870932284afd7cb934","project":"openstack/blazar","branch":"master","topic":"ci-devstack","attention_set":{},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2021-11-05 17:08:51.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-11-05 16:17:13.000000000","reason":"Pierre Riteau replied on the change"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2021-11-04 16:11:20.000000000","reason":"Mark Powers replied on the change"}},"hashtags":[],"change_id":"I5fdd5e6095e579e2584984870932284afd7cb934","subject":"CI: Fix breakage following devstack changes","status":"MERGED","created":"2021-11-04 09:38:50.000000000","updated":"2021-11-05 17:10:45.000000000","submitted":"2021-11-05 17:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816627-ci-devstack","meta_rev_id":"2b9d8ab7a9c185cd926dd3236b1aa564c9d9732a","_number":816627,"virtual_id_number":816627,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-05 17:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":2,"date":"2021-11-05 16:11:32.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-05 16:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-11-05 14:45:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-05 16:17:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-05 17:10:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 10:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 12:05:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2021-11-04 12:05:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"1449a29af0016a4f8a57590708ddc17d4c6819e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 09:38:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1386f4285d6e854e0df3b6c01e0f4e8fef25adb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 10:28:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/916010bec49b49bd9fd07938233f1abc : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43cba604d2414aab9f0bca66807aa65c : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e59eac9adbf4d469dd348cd401b7b08 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c630faba02a40428070248e37a9d5bc : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b28eda665b4aca8a98a4a705d7bc95 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92f62f1c77e34688b6494dfe798d109f : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70aa195796540adbc44c642bf8b9161 : SUCCESS in 7m 35s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b1e17699b38949648f183966de695ee9 : SUCCESS in 40m 38s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/905f2de2c416498984227ba63460374a : SUCCESS in 44m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a31e7351b3ab4461a5b2d011442d7ba0 : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99c85c27ba159170ddeaa6fdcd0f855daceb441b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 12:05:23.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7e005f801721de36bbe929a67bcb6e2719bec80","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-11-04 16:10:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"414c7c499d250d1848e1d6d10d4b78c1ecf13d95","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-11-04 16:11:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"32d3da513098e0dac5179def130af2c5ecb16b29","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 16:14:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b73b577ccee3170dcb57c844bbc734f9c7f0c8a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e27bd06edee2505c5890e6be9c5e25c819021a8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:46:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89b71e780a1f416db808392ff67fea40 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a59c36a6bc748f0bede23f804724de3 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/311c45eec7194c23856d38a1ea772e75 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc616363e5fc473ba4f24736b8c77355 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb704728648494cbeebf003b271a5c3 : SUCCESS in 10m 25s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cf6d510de501476381957ec7e0926355 : FAILURE in 28m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4760294ecf7c47e2888ca4ec9ae2aef2 : FAILURE in 26m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e7900200d1a5415c84585b7077fe64f7c5818465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 16:59:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ecf4e6e6cda44e59581cc25a20ae82b8f9cffa45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 17:34:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2154dcbaf7e48dd8f26dbdde23f6639 : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb975752bcb04e83a11a7c7dae66babd : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cb0322d89f4e4dbc1b50642971eada : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/764f1ef39087446f8c4947d98b42bf0d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3da25c10dae3456fa153bf31e9880cf6 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d95e1a5a66c242438aae0851371f243b : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c17714292540b5b6b7a2050d94b700 : SUCCESS in 10m 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/fd19b7d6606a400d8d52fa61e4fe47e2 : FAILURE in 30m 40s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b8bb9e2e972240bd8f58aeea011214aa : FAILURE in 28m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40f0863ec4ff4be2993c91ffa6d08eb7 : FAILURE in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa9ca27a224b38c041368233ea977c9003af6285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-05 12:39:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf459df4fc546fd8a8bfd72da0f15634deb4afa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 13:15:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea0003185304a30b4d256e7a3ee535d : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba6381a96109454eb1f9754ba1619c14 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea9d30243ad24900a43d460a0b6a4002 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7874c550b5fc4411858d6fc16bd6852d : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/209393159c904df7b2fa2db8c29bb737 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d64173df21c419399077d8c6257d23c : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4cec5146ee40cf8ec8a00b1b3befbb : SUCCESS in 8m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b7f67439a8fb4fa9957698fea3a3a87f : SUCCESS in 35m 20s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b23bb05802e244dd8f4d146a0ae925a9 : SUCCESS in 32m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a05df6f52e24de787da2d1ad86e8b3f : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f854df13e1d99f7f36516b7c9d5aa50f2d6d7e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-05 13:48:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"86759d784eac5831455769b859064cd296db872f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 14:33:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38e47dd98eaf4da080d9f8f5f0c66b83 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e151b8b67394c7ca0b067c7c415249b : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c123357016bd4fb4bedff26ff4808fea : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca00896b531e4334a5d763c475f409b2 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39905d7d6a634e1487245312663e1dde : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2489b2f00944775b4fd454d997e3cb0 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57af8c2c749b410d9580464193c92a3d : SUCCESS in 8m 08s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/703b9730f4524925908fceabb5063218 : SUCCESS in 39m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/71b2cc7a1066414ba40b174043743bf7 : SUCCESS in 40m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2df8aef11b24169808367666cc222ba : FAILURE in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0898dc00f3be13ef4dc57e0c4154b6db1f0c99f9","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-11-05 14:45:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0c7269eb835e8aa2f139aa9c0ec7306137b10b5d","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-11-05 16:11:32.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7fb1fc8bbc5b5de2142669aa59e3917b432fd3c7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-05 16:17:13.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b8d49fc1c4d494f25441a2a7f4089510deefdbad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-05 16:17:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"854c2214564b809aa5b97097273233870deb85ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:17:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4c95c52d83ee0bf18a270ffb7cb94eff04b19a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:08:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a5a2502517041e889159931f13fb211 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0032db944e4a2fb34915b26dbf2993 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4f6c59179d4e37b60f30787e883109 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50575592a154690a5136be23c7e8aa6 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5499af295bb3433db2a873795b568f5a : SUCCESS in 8m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/3ee958a81c654a469f4203e697be12e2 : SUCCESS in 49m 45s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/87c2bb6f525d4823813f55146584144f : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":4},{"id":"d76fe78cf66242e5f8b93a1dae4a47aaccec6c8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:08:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"92f6e381cfad0dbc204a5165aa5ab46cf01fe2d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:10:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5cf330ddf94b1db317d6057653eede : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","revisions":{"d366a5d4caa39ae1f663ddd3b98155ff380236e2":{"kind":"REWORK","_number":1,"created":"2021-11-04 09:38:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/816627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/27/816627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/27/816627/1"}}},"commit":{"parents":[{"commit":"949c52dba434d6888b14c35c4db687118f0c74e0","subject":"Ignore usage enforcement release note for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/949c52dba434d6888b14c35c4db687118f0c74e0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 09:37:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 09:37:48.000000000","tz":60},"subject":"Use KEYSTONE_SERVICE_* instead of KEYSTONE_AUTH_*","message":"Use KEYSTONE_SERVICE_* instead of KEYSTONE_AUTH_*\n\nA recent devstack change [1] has dropped all KEYSTONE_AUTH_* variables\nexcept KEYSTONE_AUTH_URI. Use KEYSTONE_SERVICE_* variables instead.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/735472\n\nChange-Id: I5fdd5e6095e579e2584984870932284afd7cb934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d366a5d4caa39ae1f663ddd3b98155ff380236e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d366a5d4caa39ae1f663ddd3b98155ff380236e2"}]},"branch":"refs/heads/master"},"32cef215fe7fda1b3b5acbf2da45136942d2e3a5":{"kind":"REWORK","_number":2,"created":"2021-11-04 16:59:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/816627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/27/816627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/27/816627/2"}}},"commit":{"parents":[{"commit":"949c52dba434d6888b14c35c4db687118f0c74e0","subject":"Ignore usage enforcement release note for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/949c52dba434d6888b14c35c4db687118f0c74e0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 16:56:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 16:56:33.000000000","tz":60},"subject":"CI: Fix breakage following devstack changes","message":"CI: Fix breakage following devstack changes\n\nA recent devstack change [1] has dropped all KEYSTONE_AUTH_* variables\nexcept KEYSTONE_AUTH_URI. Use KEYSTONE_SERVICE_* variables instead.\n\nAnother change [2] switched off the creation of an admin endpoint for\nkeystone, which we used. Switch to internal endpoint instead.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/735472\n[2] https://review.opendev.org/c/openstack/devstack/+/777345\n\nChange-Id: I5fdd5e6095e579e2584984870932284afd7cb934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/32cef215fe7fda1b3b5acbf2da45136942d2e3a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/32cef215fe7fda1b3b5acbf2da45136942d2e3a5"}]},"branch":"refs/heads/master"},"98ce96e3f0270bcfe93527494d72da6eb6990767":{"kind":"REWORK","_number":3,"created":"2021-11-05 12:39:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/816627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/27/816627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/27/816627/3"}}},"commit":{"parents":[{"commit":"949c52dba434d6888b14c35c4db687118f0c74e0","subject":"Ignore usage enforcement release note for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/949c52dba434d6888b14c35c4db687118f0c74e0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-05 12:38:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-05 12:38:36.000000000","tz":60},"subject":"CI: Fix breakage following devstack changes","message":"CI: Fix breakage following devstack changes\n\nA recent devstack change [1] has dropped all KEYSTONE_AUTH_* variables\nexcept KEYSTONE_AUTH_URI. Use KEYSTONE_SERVICE_* variables instead.\n\nAnother change [2] switched off the creation of an admin endpoint for\nkeystone, which we used. Create it again until we update Blazar to stop\nusing it.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/735472\n[2] https://review.opendev.org/c/openstack/devstack/+/777345\n\nChange-Id: I5fdd5e6095e579e2584984870932284afd7cb934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/98ce96e3f0270bcfe93527494d72da6eb6990767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/98ce96e3f0270bcfe93527494d72da6eb6990767"}]},"branch":"refs/heads/master"},"f055bbb5e92919ed2912c5fd40598ef30aedaeb7":{"kind":"REWORK","_number":4,"created":"2021-11-05 13:48:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/816627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/27/816627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/27/816627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/27/816627/4"}}},"commit":{"parents":[{"commit":"949c52dba434d6888b14c35c4db687118f0c74e0","subject":"Ignore usage enforcement release note for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/949c52dba434d6888b14c35c4db687118f0c74e0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-05 13:46:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-05 13:46:57.000000000","tz":60},"subject":"CI: Fix breakage following devstack changes","message":"CI: Fix breakage following devstack changes\n\nA recent devstack change [1] has dropped all KEYSTONE_AUTH_* variables\nexcept KEYSTONE_AUTH_URI. Use KEYSTONE_SERVICE_* variables instead.\n\nAnother change [2] switched off the creation of an admin endpoint for\nkeystone, which we need. Create it again until we update Blazar to stop\nusing it.\n\nThe devstack plugin was creating a keystonev3 service with endpoints\nusing port 5000. However, we have been using the main keystone service\non port 80 since commit 2d0cabb17d9a609a62bcffb13175e34b1c1cc1f0.\nUpdate service name/type and endpoint URLs accordingly.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/735472\n[2] https://review.opendev.org/c/openstack/devstack/+/777345\n\nChange-Id: I5fdd5e6095e579e2584984870932284afd7cb934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
