)]}'
{"id":"openstack%2Fblazar~819210","triplet_id":"openstack%2Fblazar~master~I9149eb5a046127057cc41f44fcd6b6784fcc61a1","project":"openstack/blazar","branch":"master","topic":"yoga-python-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-12-16 16:48:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9149eb5a046127057cc41f44fcd6b6784fcc61a1","subject":"Updating python testing classifier as per Yoga testing runtime","status":"MERGED","created":"2021-11-25 00:29:22.000000000","updated":"2021-12-16 16:51:28.000000000","submitted":"2021-12-16 16:48:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819210-yoga-python-testing","meta_rev_id":"0a71f806ecdc4811ba22c5e8acc9e18173128ee8","_number":819210,"virtual_id_number":819210,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 16:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-12-16 16:07:53.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2021-12-16 15:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":1,"date":"2021-12-16 16:08:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2021-12-16 16:07:53.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 01:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 15:16:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-12-16 16:07:44.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"b75532082d20c46d6732a1e857ee84a5ce91754e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-25 00:29:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99273ca51ab42ca5a27ccf90bd3e8c9932119613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:18:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6922b6b180374cd1a25e1153f3749569 : SUCCESS in 3m 23s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b64a4b6341444b083a33e7a9db4f18b : SUCCESS in 11m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d658a0279742629759f78b991036a3 : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a8ca0f4059742d39909973bcbfda62b : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc4009ead1a4cc0b90edbbbba0eb864 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5c1041c872b4653814f69f5252fe0fa : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bef3b0b2e525440d83b5fdfb63465e17 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34dfcab1fbc454982e5af2ce46e18e6 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72758b711a5b46558a827c012205e81f : SUCCESS in 4m 40s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e43d280f420f41ba8f26922abdf47cb8 : SUCCESS in 35m 13s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6456571361452197584d86419d3fad : SUCCESS in 42m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d02e6556a953462c89578b02ddecf6b1 : FAILURE in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc0a3b40a31a2ac31bbf70b9afe7a76ff2e7b2e0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-26 15:41:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5518a079fe3d27fd32ed8dfbeab7253012f18cb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-14 02:37:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0131fbce42e5f29cd18cd3a63de84d09f24b6d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 03:38:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/857f5084a4a043c2a64396bdbb678394 : SUCCESS in 3m 50s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00707f44d15243ec92b8f396899e053f : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65598a3439cd401da9bffbcebde291ef : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a7058f508f9483994423c7d4469514d : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cf18573f1047e38ab75312fcf874c9 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34a43d3a871b4ffbb447fa3108992507 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d7505776f2f47f9940b977c251f7e41 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b493231aacf40b49d51af647416e5d9 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca8d94884b5f4559a015ff972cc8d6a3 : SUCCESS in 7m 03s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/585df207ddc040738e87ddafa9b17b9f : SUCCESS in 46m 12s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/df665f0003cb4578a03950fa884f2a5a : SUCCESS in 43m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3019b3e828c4471997fb8e7cc94415e0 : FAILURE in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c7f16e7de5e049ccc2bfff53daa7f324fd38468","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-16 15:16:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d0f8be3d56979f96087579f351335ecc11e8b91","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-12-16 16:07:44.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"e40aad518a351145980fdac6065b6d787afae96e","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-12-16 16:07:53.000000000","message":"Patch Set 2: Code-Review+2 -Review-Priority","accounts_in_message":[],"_revision_number":2},{"id":"6daa581cd40af20aa387876e2e2f2c83c2184653","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-16 16:08:28.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"45d2368eeb29403094a6e2d512a0bb3d24e72f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:09:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"380339a7b9eb713bc03df60c3583f60061dc30ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:48:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9dc1866e17564e2a8ce25c53adf2307f : SUCCESS in 2m 42s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4637b784f01c49a9a5623b412ceeb3d8 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db635259420d4139ab81aa705a7a9bfc : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad07f91282a4cdbbd950ce824d57eaa : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c398e5bfaff4f40954d11ebe85b5ded : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/011db4b537d74c3e8d4a0ff76d263748 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f94a7e411f246dc835b2fb26d04b0b2 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc519725118149fa909150764dd5a4ea : SUCCESS in 4m 05s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6a362c365538466ba503a57151b159e5 : SUCCESS in 37m 05s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1ec6f94ad94b49ee8311508e07d53fbc : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2a6df5d273fa868ab5a7902f4969ce483197539e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:48:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0a71f806ecdc4811ba22c5e8acc9e18173128ee8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:51:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a0fc698cfa8f445e9859cfa126017785 : SUCCESS in 1m 35s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fa671205ea43daa2db618d660e0fbe : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6afd4d326c1b4c29a7e035ec639fa4df : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d85abc7589482dea6bc202606a98226776fde54a","revisions":{"c3e55044c2f65e9761a489d57adf4e4a71e90105":{"kind":"REWORK","_number":1,"created":"2021-11-25 00:29:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/819210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/10/819210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/10/819210/1"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:26:39.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:26:39.000000000","tz":-360},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime has been updated with py38 and py39\nas voting and removed the py36 testing. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/818609\n\nthis commit makes other required changes in setup.cfg metadata.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I9149eb5a046127057cc41f44fcd6b6784fcc61a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c3e55044c2f65e9761a489d57adf4e4a71e90105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c3e55044c2f65e9761a489d57adf4e4a71e90105"}]},"branch":"refs/heads/master"},"d85abc7589482dea6bc202606a98226776fde54a":{"kind":"REWORK","_number":2,"created":"2021-12-14 02:37:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/819210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/10/819210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/10/819210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/10/819210/2"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-25 00:26:39.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-12-14 02:37:22.000000000","tz":0},"subject":"Updating python testing classifier as per Yoga testing runtime","message":"Updating python testing classifier as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I9149eb5a046127057cc41f44fcd6b6784fcc61a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
