)]}'
{"id":"openstack%2Fblazar~820255","triplet_id":"openstack%2Fblazar~master~I47424bd24b894669381b16fbbd401cf58acc590c","project":"openstack/blazar","branch":"master","topic":"bp/rpc-refactor","attention_set":{"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2021-12-02 20:36:44.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2022-06-10 16:15:34.000000000","reason":"Jason Anderson replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-07 14:04:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I47424bd24b894669381b16fbbd401cf58acc590c","subject":"Remove RPC layer from API","status":"NEW","created":"2021-12-02 19:47:43.000000000","updated":"2023-03-25 20:04:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"30a49211af212e68e61781146cb217d3d8e2d56c","_number":820255,"virtual_id_number":820255,"owner":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-24 17:07:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":-1,"date":"2022-06-10 16:15:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-07-07 14:04:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:36:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 23:02:04.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-28 23:02:04.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2022-08-25 09:02:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-03-25 20:04:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"}],"messages":[{"id":"e2375a0d57f0edea1813adc4c5c0a8fae5b7033a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-02 19:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e176bbaec706d4594132ff7d8a231b36345ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:36:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbfecd4e99c645d597e4d6d8135daef0 : FAILURE in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64f58009a8b1457fa85c62b14bc30f1c : FAILURE in 6m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc01c28147640a3b637e07651b54314 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f933dd8050e48b0a6f9eefc27949949 : FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eacb75a3ae5c45468117d54bca6ec19e : FAILURE in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1df2640ed514f6c90a9efcd8e6f7df2 : SUCCESS in 9m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/99217c0f713145208037ba992243b18f : FAILURE in 41m 16s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/2ad1b86c3def4bf3b4857504567339d8 : FAILURE in 35m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/783dc0fc7dc14510adc90ec8e95c7420 : FAILURE in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae45e3530f5ab9ed57508bfa3767461665529e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-07 16:09:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af399b70cd4871606c89055696b0c589ac1860b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 17:01:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0f02e5f89d149c5a43a8a06d6d4d8f8 : FAILURE in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/754c6f3892994d2995acfda109eaf8b5 : FAILURE in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68cc811b2be44cb3a5fea41522f2d671 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b07cfdb974cd455ab05376818b5a7872 : FAILURE in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6bdb813493041a2be45620e9285a72e : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b53b073a154f268a17bf330fe9cd38 : SUCCESS in 6m 33s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/349a105b273c4b06813bceb8f13e6e43 : FAILURE in 34m 17s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9dbb8545be9b4518918c80fad95231fb : FAILURE in 39m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d8838d2a6254e5f9d247f2530bc7f6a : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b773dee69ae1dfefa4593751cc4feb9a1817e8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-08 16:20:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45226ad76bc353455f990cab6ad9cfb6aafc4fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:55:34.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6815ac7329e048148357967e8d0af231 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f98a1167a84643b1bfb5a6008bc40527 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902ac4da1bb547edad9f44510123e77f : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ade7096fd1d480ea623cc56b8fa50ca : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b6de3c715984d15a443e5e30b3df8e9 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2209bd5d061f4349bc1bcde14a5d723f : SUCCESS in 7m 56s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2dfbab05c6734cdb9abeee82f009ca1c : FAILURE in 30m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/303f9dc5598743e08bac5ffd63c8e8a6 : FAILURE in 31m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94d40dc5e8dc4f2abf6da644de4e7746 : FAILURE in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6536ee4a67e853c8d4f25b77dde08eaa56f9b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-08 17:17:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa974afe160cd3384d3fae2c437ec3e46e76bc5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 17:51:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4c96670ffb1489387f0fe354440f3e5 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94c7576ccd14494e9ee0215c679543d2 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c004f38a624327a063ee8a7fa73f31 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e1c7fe636a24caf8fb9cb414f058036 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a3d07485aa54e7087d80dfd442f0cbd : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e18e1d750447ce98cc4746e61c6fa7 : SUCCESS in 8m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b4bcf842d20c4a858ad2085c85430961 : FAILURE in 31m 20s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1b4c0657c4f94e499739b9a88dba2794 : FAILURE in 30m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e151a15fe6b493f91b3e16ff4189401 : FAILURE in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb154e4073439c7664626ce4c4fa4212c9964476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2021-12-17 22:36:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7ebbbb06303c4881d1c7d29b736836bc35b27afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 23:14:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a9857abb4724276a358cc0dbfc34138 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/027889e8a4334babb218eeaa26c732bc : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710f4591c4a74cdc95fdd6f5b4066c2d : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe9a639736594b1dba8089ebc3bb5af5 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53fa35b166db49028fa9a447d21c278e : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba2d0d8438a4d7f9143031c60b7c94a : SUCCESS in 7m 10s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/bcaa12e8a326472fafbfadf7bf4eade6 : FAILURE in 33m 39s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3760997c411e43048e96d3783f7d5838 : FAILURE in 33m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75dd23fcaf494851a950f4c27d9ddf50 : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5ae0e7c938c970194921e8780aae246a05321b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-04 22:36:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"19aba5a9e28a81cf3143bdbe9096f54f00d4dc1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 23:10:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7425f13c5534a02a18ee44a0bd4db42 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eef525f62ee84ab0a0f1b4b6dc6d3837 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2462d2d5e0e041eca0c2338711fd7f16 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb936a2d7960452684757b1d763b97e1 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/198019a1f94f45ea9d997180f7ceaf4b : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f4265923eb428bb1a6acc8bd69d0b9 : SUCCESS in 7m 19s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4c66c0553bf8449a92eb2c20cc03fdde : FAILURE in 32m 43s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f8e35a05adeb47b18bd71139348ee5d8 : FAILURE in 29m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cbbec10451054dda8727b0e631cf682b : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0784286495654405dfb7508aa30af802a37774b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-11 22:34:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a2e1f1650cba51cc989c82e1bfe7e911608349d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 23:17:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9b01b2ff554b51865330709db01c9e : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/970afc9c1990403482a2e31e066724a7 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3580acdcf2e846daa4592d32d1396969 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/827cba9ce50a4abca6b541f154da2349 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d19e0ebab4604169b9f27006e49e7fa0 : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58d376dadcb409185f8088ba30307f2 : SUCCESS in 8m 01s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/de3bcaf8fabe4522a0a99ac128047627 : FAILURE in 27m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c44cd4eaf09c4422a772297ac20e21db : FAILURE in 37m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ab06bc41add45c682125edb3a3c225c : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7413ea285fa815ad8305320611e0940b81d89b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-24 23:00:49.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"1d3af7c6d6a1ce0f3fcac1e9318db49ee21ade4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 23:36:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4df44eda0a4949079dc7213e9f140ace : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee3a94b4c0114f0c831b2d208363f410 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4168250d75d345c49d07cc5859e65421 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5147c9cfaa974841b0358e12ef45e02b : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cffbcf5b9ba434fbe98e4f538a31f65 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f369795db5304bf093f706792ddd21cb : SUCCESS in 6m 52s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6c16730232e54884b213ba74c9877d98 : FAILURE in 29m 00s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4f0dbd37e36748c68186cec7dad657dd : FAILURE in 31m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c81317898014c1789c4d6ce3b6c5441 : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b62f1256a53d166a20a0472b872222ccfa88b889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-25 17:51:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8dcc1930497e154d6635038ddd1b289cdb37dc49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:24:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f0baa6d5e0e425eb5470e3230a1acab : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5724a09f3d0454381d7b20db819915c : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f090828db0c34b01a84f145df266c33a : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72969b93c44942b7a8fd9f3d35e5f60c : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89d8729d340f4f00a7849af60b937a9d : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c906bdd4a349eaa9d521a7db655b98 : SUCCESS in 9m 32s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/eba279b862364d7182565e44fd8fd5f6 : FAILURE in 25m 55s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c43feebf8fea4019b3d3d47825aef59e : FAILURE in 26m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b32c59118b10405f89ede5cfa85f380b : FAILURE in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"162e9acee2feeea940c47f2845cc4914dc76dff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-28 17:33:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"640fb1541519ae75df1a68fea22f0346a7f3e30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 18:50:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/285f2b4efd2847d8b095fa6f1a9410ed : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6129917a84c34aabaa82841012c58233 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabb61d4979e447bb3561c8d2dba4574 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6072a041c04d69b73af0bb7360a568 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fe83c0945dd492496e95a77c7bbd3f4 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a476921281499791b941202dc02000 : SUCCESS in 10m 40s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e747fa86d9434107a8c9a7ff03df9362 : FAILURE in 40m 46s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/eb53583d26cf4e8c8f9d549d8f79e690 : FAILURE in 40m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff2fe711ae5e42ebaacdc8b4e82cb28a : FAILURE in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6119ff1e07d496bea3091fc38d4029a67e63ed97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-01-28 22:42:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e5aba56585f799d0f0444b43cdda23186123c5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 23:16:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d35b895a6fed4166aea7dedf465a6624 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8644e3cf104b43698b04cccef8badb1e : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44df28a76f6048f89d317b130f37094e : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5bd215a483e4d1eaf085ee4c9e57f96 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b96ea1ee46a4b7b92d14550fc2791db : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765a121d3f1340aab66857be52683330 : SUCCESS in 7m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f99cb4c7921c4241b0b19520039720cb : SUCCESS in 30m 51s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc2c076ca8b4b559fcc5df3e276b9f2 : SUCCESS in 27m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91740f04e7fd4e36b279ab412ca11206 : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"54ae6470024c07d0962a31e6af137a267493b3fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-02-24 16:12:27.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"6e4039ffd962b6d8b5567254618f009d969c8149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:07:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/287c01600e5d4f25ba4b30e1a05b9cff : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/478cda1e999246929a01b8ee5fc1041f : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2577dfff77734da880bc2d16c339f0f7 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8fd84a1c07c4b97aee6182d980524ab : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73dfeb32e2804bc9bd0de24b4735046f : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c89b343059e4d6f9206446e0ec26e9b : SUCCESS in 7m 12s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/358fff03975a4ab78a0ae19bbed921a7 : SUCCESS in 46m 12s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/35f4e3553f7e4aa3b78e3c6ba820f49c : SUCCESS in 49m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96b387189b7a4529adc34c0c56474386 : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"443305265c48911f4db4d6d5b2eafd1d6100b725","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-06-10 16:15:34.000000000","message":"Patch Set 12: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8bbd400a50255711252735f8d0acb75d3048338c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-07 14:04:35.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"363388e9afa1bfc56a64f7bbe958c07bb223cc45","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-25 09:02:07.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"30a49211af212e68e61781146cb217d3d8e2d56c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:04:41.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e with the following votes:\n\n* Code-Review-1 by \u003cGERRIT_ACCOUNT_30491\u003e\n","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":12}],"current_revision_number":12,"current_revision":"718baab60bf564d3aa7901e961db7f9a437a52ae","revisions":{"e8c7626e53cf7b455323b0dc134b8f63d1b0aaa0":{"kind":"REWORK","_number":1,"created":"2021-12-02 19:47:43.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/1"}}},"commit":{"parents":[{"commit":"7288e66c2cd0d4969d14d335c6eaa12076be0c63","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7288e66c2cd0d4969d14d335c6eaa12076be0c63"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e8c7626e53cf7b455323b0dc134b8f63d1b0aaa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e8c7626e53cf7b455323b0dc134b8f63d1b0aaa0"}]},"branch":"refs/heads/master"},"f5e4d0b6d7dae4657b0cfe27e4da6768a44dcdfe":{"kind":"REWORK","_number":2,"created":"2021-12-07 16:09:08.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/2"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-07 16:08:53.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f5e4d0b6d7dae4657b0cfe27e4da6768a44dcdfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f5e4d0b6d7dae4657b0cfe27e4da6768a44dcdfe"}]},"branch":"refs/heads/master"},"b85d522b36d6747cb5a31ae14d770a8f5d11dbe6":{"kind":"REWORK","_number":3,"created":"2021-12-08 16:20:13.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/3"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-08 16:19:52.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b85d522b36d6747cb5a31ae14d770a8f5d11dbe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b85d522b36d6747cb5a31ae14d770a8f5d11dbe6"}]},"branch":"refs/heads/master"},"08a69b76c73ec498ee6521c31baeb59e3018ffc6":{"kind":"REWORK","_number":4,"created":"2021-12-08 17:17:14.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/4"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-08 17:16:57.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/08a69b76c73ec498ee6521c31baeb59e3018ffc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/08a69b76c73ec498ee6521c31baeb59e3018ffc6"}]},"branch":"refs/heads/master"},"2a7c1e708d2d59758b7f8d71774dd2fce50a03e8":{"kind":"REWORK","_number":5,"created":"2021-12-17 22:36:50.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/5"}}},"commit":{"parents":[{"commit":"f055bbb5e92919ed2912c5fd40598ef30aedaeb7","subject":"CI: Fix breakage following devstack changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f055bbb5e92919ed2912c5fd40598ef30aedaeb7"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-17 22:31:33.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2a7c1e708d2d59758b7f8d71774dd2fce50a03e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2a7c1e708d2d59758b7f8d71774dd2fce50a03e8"}]},"branch":"refs/heads/master"},"185f9224d0f22da311c0da80ffd041137787270b":{"kind":"REWORK","_number":6,"created":"2022-01-04 22:36:05.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/6"}}},"commit":{"parents":[{"commit":"a55e954bd55e53d3df4d839d99c12f166faa71d9","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a55e954bd55e53d3df4d839d99c12f166faa71d9"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-04 18:16:09.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/185f9224d0f22da311c0da80ffd041137787270b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/185f9224d0f22da311c0da80ffd041137787270b"}]},"branch":"refs/heads/master"},"e8e68b9c1a25d38ee8432243dea09323706b4856":{"kind":"REWORK","_number":7,"created":"2022-01-11 22:34:40.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/7"}}},"commit":{"parents":[{"commit":"10072b1e68327f8bb03a6064d2ea2fa91fa4ace1","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/10072b1e68327f8bb03a6064d2ea2fa91fa4ace1"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-08 21:43:57.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e8e68b9c1a25d38ee8432243dea09323706b4856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e8e68b9c1a25d38ee8432243dea09323706b4856"}]},"branch":"refs/heads/master"},"5ddba6343ab282c661205eb6773c780ebd7516f8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-01-24 23:00:49.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/8"}}},"commit":{"parents":[{"commit":"20ba175d1c2ce3f93f905d11cd116c8c3d07be35","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/20ba175d1c2ce3f93f905d11cd116c8c3d07be35"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-24 23:00:49.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5ddba6343ab282c661205eb6773c780ebd7516f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5ddba6343ab282c661205eb6773c780ebd7516f8"}]},"branch":"refs/heads/master","description":"Rebase"},"650e6f6d9c4e36cc249edeef1a916a4927e81da4":{"kind":"REWORK","_number":9,"created":"2022-01-25 17:51:32.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/9"}}},"commit":{"parents":[{"commit":"21c020ea2ed022e999aa5d3cbfeaef5210162c2e","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/21c020ea2ed022e999aa5d3cbfeaef5210162c2e"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-25 16:31:34.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/650e6f6d9c4e36cc249edeef1a916a4927e81da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/650e6f6d9c4e36cc249edeef1a916a4927e81da4"}]},"branch":"refs/heads/master"},"1b9f857771084dc04a334771f8a0bf5c567588f1":{"kind":"REWORK","_number":10,"created":"2022-01-28 17:33:31.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/10"}}},"commit":{"parents":[{"commit":"75ec9d7072e550bb4965fb619f60ccb0b54c1fd4","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75ec9d7072e550bb4965fb619f60ccb0b54c1fd4"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-28 17:32:44.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1b9f857771084dc04a334771f8a0bf5c567588f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1b9f857771084dc04a334771f8a0bf5c567588f1"}]},"branch":"refs/heads/master"},"a9667af666194f35809d25b9d9396661fb23a01d":{"kind":"REWORK","_number":11,"created":"2022-01-28 22:42:51.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/11"}}},"commit":{"parents":[{"commit":"75ec9d7072e550bb4965fb619f60ccb0b54c1fd4","subject":"Add resource properties discovery API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/75ec9d7072e550bb4965fb619f60ccb0b54c1fd4"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-01-28 22:42:43.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a9667af666194f35809d25b9d9396661fb23a01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a9667af666194f35809d25b9d9396661fb23a01d"}]},"branch":"refs/heads/master"},"718baab60bf564d3aa7901e961db7f9a437a52ae":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-02-24 16:12:27.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/55/820255/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/820255/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/820255/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/820255/12"}}},"commit":{"parents":[{"commit":"e9c19e2206839d4f73bb94aab579fbec596df8f4","subject":"Merge \"Prevent conflicting events from running concurrently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e9c19e2206839d4f73bb94aab579fbec596df8f4"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2021-12-02 19:47:23.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2022-02-24 16:12:27.000000000","tz":0},"subject":"Remove RPC layer from API","message":"Remove RPC layer from API\n\nThis removes the RPC layer from the API, and instead calls the\nplugin functions directly.\n\nImplements: blueprint rpc-refactor\nChange-Id: I47424bd24b894669381b16fbbd401cf58acc590c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/718baab60bf564d3aa7901e961db7f9a437a52ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/718baab60bf564d3aa7901e961db7f9a437a52ae"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
