)]}'
{"id":"openstack%2Fblazar~824579","triplet_id":"openstack%2Fblazar~master~Id7771b63116064f2adc0cdfd700f14afa1ea6170","project":"openstack/blazar","branch":"master","topic":"bug/1957761","attention_set":{},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2022-02-04 15:54:55.000000000","reason":"Jason Anderson replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-04 16:49:05.000000000","reason":"Change was submitted"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2022-02-03 14:50:41.000000000","reason":"Mark Powers replied on the change"}},"hashtags":[],"change_id":"Id7771b63116064f2adc0cdfd700f14afa1ea6170","subject":"Fix lease update when resource_type parameter is missing","status":"MERGED","created":"2022-01-13 14:26:11.000000000","updated":"2022-02-04 16:51:58.000000000","submitted":"2022-02-04 16:49:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824579-bug/1957761","meta_rev_id":"1af540aa53be5039781b393505dbac57b6a318d1","_number":824579,"virtual_id_number":824579,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 16:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":2,"date":"2022-02-04 15:54:55.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 14:50:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":1,"date":"2022-02-04 15:54:55.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 15:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 12:10:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2022-02-03 12:10:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"589403ec67a2a71a3e3efc2dee960178d47e6c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 14:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fae79b07595cd246076d28964b38ec43d01bb5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ad537d5065a411ab09d304c4a37c128 : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/882d208ef8444fadb50286bc3ff4900c : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c659a2cb8554d1f9b343fe24c8f0009 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83eaf37738474c8e8d4aadd8fc33ff8b : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e875eb0485840b096f280cd13c3ebf7 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19aaf1ef1d29413d9343574047d871c2 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c049e1452648d1bd9d5f0f3e0dff74 : SUCCESS in 6m 42s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/9eb3dc0a62a0485cae3ef5de7215266e : SUCCESS in 45m 01s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ea9fddf1667b414f9475818922dbeb51 : SUCCESS in 43m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/111cddc1be9b4e9ead563b0d16a22cf0 : FAILURE in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc60d0695a491002dac4b5914d15dac40362de9b","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2022-02-03 14:50:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffb21fd9d8636b394e74140be00f7a74dce1648f","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-02-04 15:54:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"514449ea00f78d15d3f0607c8de4825a19c5cd41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 15:56:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6e15661c82574c0eed031d1252c9395f3e80597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:49:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4600365d066a44b189947d567ee81103 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26965e764e6246659ee604a01ec4a847 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab0946ee64224e76a178040ddf923c66 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db1db7ac93d64f09b51dc96440b4c673 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7146a93b3c04542bd1873a57a06fe20 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46a40bc81df648dab6f9577a4068aa51 : SUCCESS in 4m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/bdce00c8573841b783e16e893f4a4bd8 : SUCCESS in 45m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/87335476652741759e06c6901daabd8f : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0f911f8d8cc2b708accaa0674cb267e929160bb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:49:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1af540aa53be5039781b393505dbac57b6a318d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:51:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa0b35aec884c6a9d713dba59a01494 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f427e095ff4f588367f77e9de11b5a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"caf4b282f046d1f50d46eb657f575fda1a4fdae3","revisions":{"caf4b282f046d1f50d46eb657f575fda1a4fdae3":{"kind":"REWORK","_number":1,"created":"2022-01-13 14:26:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/824579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/79/824579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/824579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/824579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/79/824579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/79/824579/1"}}},"commit":{"parents":[{"commit":"d85abc7589482dea6bc202606a98226776fde54a","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d85abc7589482dea6bc202606a98226776fde54a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-13 14:23:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-13 14:23:52.000000000","tz":60},"subject":"Fix lease update when resource_type parameter is missing","message":"Fix lease update when resource_type parameter is missing\n\nChange-Id: Id7771b63116064f2adc0cdfd700f14afa1ea6170\nCloses-Bug: #1957761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/caf4b282f046d1f50d46eb657f575fda1a4fdae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/caf4b282f046d1f50d46eb657f575fda1a4fdae3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
