)]}'
{"id":"openstack%2Fblazar~825255","triplet_id":"openstack%2Fblazar~master~Id847acd613c27d4e647391c346ab6d3ceced99e4","project":"openstack/blazar","branch":"master","topic":"log-warn","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-27 16:59:38.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-19 09:58:19.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Id847acd613c27d4e647391c346ab6d3ceced99e4","subject":"Use LOG.warning instead of deprecated LOG.warn","status":"MERGED","created":"2022-01-19 05:18:11.000000000","updated":"2022-01-27 17:13:21.000000000","submitted":"2022-01-27 16:59:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825255-log-warn","meta_rev_id":"789560cf62ba66613cf097081881278116409ebb","_number":825255,"virtual_id_number":825255,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 16:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 16:08:18.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2022-01-19 09:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 16:08:18.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 05:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-19 08:48:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-27 16:08:18.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"c4cdfec60e03f6cadf3a9d4f3535c5f033f9cf80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 05:18:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d990131fa64f7d950a0d8ccd69d9b3506b5cc0dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 05:58:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67e1e3c290214e6fa5d3f94c05ff8ce6 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c31d46405344f6296072b39e26a368d : FAILURE in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ede120c9d9d4c608f2f441898443698 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524e8a3d91164a69a7ee29efb215ea9d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d04c00437994222a73639b9394add45 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/469b535f4acd4b88b0d377e1cd2fdeec : FAILURE in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aac77c5e06d4690a2e68e1fd401c2a3 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1627da866bec432c82b1badffefee496 : SUCCESS in 5m 10s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5f11fd2bedaa48a5994eced417db613c : SUCCESS in 36m 57s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0955403d887941a399a8f803bf253b95 : SUCCESS in 36m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abececa9e430448eafc00094bb11c74b : FAILURE in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fae43a8252e51c8a5336cedc9145f355bbc975e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 07:28:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0bd8f028e07cd0610b91908f5a30b6401e9927e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 08:07:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd2211e07424498c96e4b3050b959935 : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d280b544e7e49e4a0943d0a46d0a4be : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f14ae95aa8b4060bbd3129479169985 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1696f6ce673643cda3b12a79c3bf960f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c1cab30a2714a5abef6d6d5db1a746e : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d0e99b2c6bc4c35ab9dba38bd040fe6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3856f2d4666c41d7bdd885ccf77b2dde : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d89b7eccad44ed0a20c2a8668cef852 : SUCCESS in 5m 31s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/135345f54cdb4212aa2f3c708e631a67 : SUCCESS in 33m 14s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a16c87f1d4e74ec8beae319ae25868b9 : SUCCESS in 36m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dad246be0dcc4b5bbb2d5ec87fad89d2 : FAILURE in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67fcc0d5ed1479e4623fb21d1fd79f4c8c2c5f69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-19 08:48:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fbb6d8e459042f0005f3f267e5e5e11d7cb50a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 08:49:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4b64c9ec7264bdf5c3f64dbb396dfcc7fdb5a71","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 08:50:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"177b2401d04a5f3d111d2e4d7cde0eff552348f2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 08:52:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2da835afd3d7903b4a83441e7a58fad515727c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 09:37:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a422e9693de5465499231db42337ce87 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0e83d6049746e79a450acf9b9c80a1 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2accd1ca75244b7b9034929148d0ce86 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab4dd16d16f412b88dd87efaaa0fb75 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5945d05ec5d4ed68e6134f886ec0481 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdfd24c6547542dcae3cfa673998b748 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155dc741c5e24179b1f0926e7cf10260 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b0d79fafbd42219cefe06510089fd0 : SUCCESS in 4m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/abd48e9661114404a0e57cc1823f9c2a : SUCCESS in 40m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/7cef58bdafc34fea9511b75c40d635d2 : SUCCESS in 41m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0c2cc2459d446d0902f37c7ef17dbf3 : FAILURE in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f015d2882c5b46eb7c942eec68ff46e88cbe786","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-19 09:58:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9fb3a79bc8772578c59ac6093b0a80f9337e35fd","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-01-27 16:08:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4409a752519864d02b1c5ab23d5c767d52fe51b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:11:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07ec1684d4df415cc21ed6ae00badc878b5b3fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:59:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a125aea19afb4ba6b89612b84cc3f9ff : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9747c65120744959886cbe521c4a654e : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b6235b7e2546e79097a533dfb741af : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf88e653e70247b1b03a1278f978d06c : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01092ee76b1746b089f08ea2ef50eac9 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c9b57745e94028a4aa705aa323307c : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd89d2266ea4380b6ce72a7e847ab9d : SUCCESS in 4m 36s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/90a6283e57714ce287004209f385ad9c : SUCCESS in 31m 25s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4fa37badf05b4068842403eaf8939ece : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"0bae87c15a3e88f2e4b0e75015b85c76814f570c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:59:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"789560cf62ba66613cf097081881278116409ebb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:13:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/667eead2d5ca43a88ac6b849f39a4daa : SUCCESS in 1m 18s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92220ffd49454428937598a4ee21aa71 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e771678603b04cc9bf69e7244ce84198 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86facea1e4f55b8fb70b85de6f8f1a24e162c4e6","revisions":{"d720e4013fd00bf0f9496e327220123c331e44c8":{"kind":"REWORK","_number":1,"created":"2022-01-19 05:18:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/825255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/825255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/825255/1"}}},"commit":{"parents":[{"commit":"39c67f09b8095c4d8aadfe550f0aa924a3693f5a","subject":"Fix list_allocations for multi-host reservations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/39c67f09b8095c4d8aadfe550f0aa924a3693f5a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 05:17:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 05:18:08.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Id847acd613c27d4e647391c346ab6d3ceced99e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d720e4013fd00bf0f9496e327220123c331e44c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d720e4013fd00bf0f9496e327220123c331e44c8"}]},"branch":"refs/heads/master"},"10a9113a37fdafd5ec8d132e1e56dd7841fb92cd":{"kind":"REWORK","_number":2,"created":"2022-01-19 07:28:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/825255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/825255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/825255/2"}}},"commit":{"parents":[{"commit":"39c67f09b8095c4d8aadfe550f0aa924a3693f5a","subject":"Fix list_allocations for multi-host reservations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/39c67f09b8095c4d8aadfe550f0aa924a3693f5a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 05:17:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 07:28:46.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Id847acd613c27d4e647391c346ab6d3ceced99e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/10a9113a37fdafd5ec8d132e1e56dd7841fb92cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/10a9113a37fdafd5ec8d132e1e56dd7841fb92cd"}]},"branch":"refs/heads/master"},"86facea1e4f55b8fb70b85de6f8f1a24e162c4e6":{"kind":"REWORK","_number":3,"created":"2022-01-19 08:49:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/825255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/55/825255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/55/825255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/55/825255/3"}}},"commit":{"parents":[{"commit":"39c67f09b8095c4d8aadfe550f0aa924a3693f5a","subject":"Fix list_allocations for multi-host reservations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/39c67f09b8095c4d8aadfe550f0aa924a3693f5a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 05:17:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 08:49:31.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Id847acd613c27d4e647391c346ab6d3ceced99e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/86facea1e4f55b8fb70b85de6f8f1a24e162c4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/86facea1e4f55b8fb70b85de6f8f1a24e162c4e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
