)]}'
{"id":"openstack%2Fblazar~828771","triplet_id":"openstack%2Fblazar~master~If20c20f6cfb6aa23cb6e19020301bf59044aa79c","project":"openstack/blazar","branch":"master","topic":"devstack-keystone-public","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-21 22:00:59.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"If20c20f6cfb6aa23cb6e19020301bf59044aa79c","subject":"Use configured endpoint type in url_for","status":"MERGED","created":"2022-02-10 17:19:33.000000000","updated":"2022-02-21 22:56:21.000000000","submitted":"2022-02-21 22:54:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828771-devstack-keystone-public","meta_rev_id":"c6e6171624e51c9c468ffe2c55176a15750153af","_number":828771,"virtual_id_number":828771,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 22:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-11 15:54:46.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2022-02-21 22:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":1,"date":"2022-02-21 22:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 18:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 15:54:46.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"ec619ae627e0086a767282bcbfdb272ca9a482e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 17:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca02f9e9879a3e9ad5e6c8b0f1f3a3cff8852399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:16:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61d90edde1934ec19e2dbd35d6626907 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c23aeeb75c6e4bea951630cef9e81158 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cac011f8be646feb3ca89f299e495f7 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/816ff60b7bf645f682a72b54bf37e8df : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff35f46e46c447baa91a508daea44ddb : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5220b7d86fb74c5a8423c2a87a4775f3 : SUCCESS in 7m 40s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/933434ad08c94675afd7fc5cba5bab8b : FAILURE in 34m 19s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8cead1558f4542b5aa338855fae8bff1 : FAILURE in 36m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/176e507117f446b0919bcc3e53b373f9 : FAILURE in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a65f07c330a6e7761be843a96c972402369d5a4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 21:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d06f328818d150e46193cdb28a4b002b461f57a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 22:48:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f84345c92e84498197c185ed5699d61b : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45336d342dcc4cf492b2c6bcd29f8abc : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975b7a1f44db40b99f64cae8c57c80b1 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99f53b95279e42a3a1d564f21e349a4f : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5fa8777a7324395ac9828c988c3f235 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4324d66463d4ce2ab2465353ab440a1 : SUCCESS in 7m 58s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/147fe00f0132469b9ec1192e4cc0e339 : FAILURE in 29m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/86a29b76c904424caf82cde761b81d83 : FAILURE in 45m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4aa0db33b474c4499b1ce2b32da60cb : FAILURE in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47c712d908785d26457a27564f39636486d4bbf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 10:02:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fd3b403406d6052e62773342882ffe20db4028e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 10:48:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d9cb02a699149aea0573af38634bfe1 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9daade19c3e4494acb80eb6b6cdfa56 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df29c9cdc19a4f3eaf63453d0a8529ac : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27eadfc7f4fa468cb9e8982f13475880 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61defb499f9a4658880c0c28dc61f5fc : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924c11d0f98d4e3fa8837d4db5d8065b : SUCCESS in 6m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/53326dc42798435bafa7f1246ae53fb6 : SUCCESS in 40m 37s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4c358afd1b724fc5a95134dfcb8c0d03 : SUCCESS in 39m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f691cb772464ba893bc7670e9f74da2 : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47a935fb25960954ddf7c17c8e5e9e64625ef4f8","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-02-11 15:54:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fb3deab378a1eb015720a74aea7613a505302fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:00:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71607f954c5520da216a72853cd6158bc2fe96d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:03:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20e409baba6fac169fb1c83c12f23cc36f418794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:54:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3015629db4b4a66997a47925720f69e : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7710f4019514a5c873e13e893a75453 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ee718c8053241a98d9960ef44e399b0 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b194d34b66a44d78ff956fc96e033ce : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef616b6ef612475cb8c5a72e9757b533 : SUCCESS in 7m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8733615015234daf99c0174656bcbdac : SUCCESS in 46m 30s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/131425152a174367b849689be895b8e8 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5e7ce78fff0717337f86bead283a1fc7c18be85f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:54:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c6e6171624e51c9c468ffe2c55176a15750153af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:56:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde3f4a55b3944b5bcb1ae55c6adeb33 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"34b848aff2363e8c72a0b0f0f2d37d0ac415a2f5","revisions":{"18162ea27a106be87250414b0bb6f343c1c14908":{"kind":"REWORK","_number":1,"created":"2022-02-10 17:19:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/828771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/828771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/828771/1"}}},"commit":{"parents":[{"commit":"9a37ded91964cf68eb1c8572c07618e2f19120ac","subject":"Merge \"Switch to using internal endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9a37ded91964cf68eb1c8572c07618e2f19120ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 17:18:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 17:18:54.000000000","tz":60},"subject":"Use keystone public endpoint for devstack","message":"Use keystone public endpoint for devstack\n\nAs explained in [1], a recent devstack change switched off the creation\nof an admin endpoint for keystone. Now that we support configuring\nendpoint types, switch to using the public interface for keystone.\n\n[1] https://review.opendev.org/c/openstack/blazar/+/816627\n\nChange-Id: If20c20f6cfb6aa23cb6e19020301bf59044aa79c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/18162ea27a106be87250414b0bb6f343c1c14908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/18162ea27a106be87250414b0bb6f343c1c14908"}]},"branch":"refs/heads/master"},"a133915f8868d3ed9d91d8f4a980cbbe4c944560":{"kind":"REWORK","_number":2,"created":"2022-02-10 21:59:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/828771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/828771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/828771/2"}}},"commit":{"parents":[{"commit":"9a37ded91964cf68eb1c8572c07618e2f19120ac","subject":"Merge \"Switch to using internal endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9a37ded91964cf68eb1c8572c07618e2f19120ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 21:58:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 21:58:54.000000000","tz":60},"subject":"Use keystone public endpoint for devstack","message":"Use keystone public endpoint for devstack\n\nAs explained in [1], a recent devstack change switched off the creation\nof an admin endpoint for keystone. Now that we support configuring\nendpoint types, switch to using the public interface for keystone.\n\nAlso replace use of a hardcoded endpoint type by the corresponding\nconfiguration option.\n\n[1] https://review.opendev.org/c/openstack/blazar/+/816627\n\nChange-Id: If20c20f6cfb6aa23cb6e19020301bf59044aa79c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a133915f8868d3ed9d91d8f4a980cbbe4c944560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a133915f8868d3ed9d91d8f4a980cbbe4c944560"}]},"branch":"refs/heads/master"},"34b848aff2363e8c72a0b0f0f2d37d0ac415a2f5":{"kind":"REWORK","_number":3,"created":"2022-02-11 10:02:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/828771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/828771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/828771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/828771/3"}}},"commit":{"parents":[{"commit":"9a37ded91964cf68eb1c8572c07618e2f19120ac","subject":"Merge \"Switch to using internal endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9a37ded91964cf68eb1c8572c07618e2f19120ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-11 09:58:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-11 09:58:48.000000000","tz":60},"subject":"Use configured endpoint type in url_for","message":"Use configured endpoint type in url_for\n\nThe url_for function was defaulting to the public endpoint if no\nspecific endpoint type was passed as argument.\n\nAlso switch from admin to internal keystone endpoint by default and\nreplace use of a hardcoded endpoint type by the corresponding\nconfiguration option.\n\nAs explained in [1], a recent devstack change switched off the creation\nof an admin endpoint for keystone. Now that we support configuring\nendpoint types, switch to using the public interface for keystone when\ndeploying blazar with devstack.\n\n[1] https://review.opendev.org/c/openstack/blazar/+/816627\n\nChange-Id: If20c20f6cfb6aa23cb6e19020301bf59044aa79c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/34b848aff2363e8c72a0b0f0f2d37d0ac415a2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/34b848aff2363e8c72a0b0f0f2d37d0ac415a2f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
