)]}'
{"id":"openstack%2Fblazar~833113","triplet_id":"openstack%2Fblazar~master~I77fb08a3811ed1445e31b3552ca23d5bfc43a07d","project":"openstack/blazar","branch":"master","topic":"enforcement-docfix","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-05-06 07:44:12.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I77fb08a3811ed1445e31b3552ca23d5bfc43a07d","subject":"Fix references to start and end dates","status":"MERGED","created":"2022-03-10 15:48:07.000000000","updated":"2022-05-06 08:35:53.000000000","submitted":"2022-05-06 08:34:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833113-enforcement-docfix","meta_rev_id":"c27b19369f80e3677eee51552a87ec8f1f983504","_number":833113,"virtual_id_number":833113,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 08:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-06 07:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-06 07:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 16:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a34079c0c35ce79d6b346084a61e59bb29a5228f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-10 15:48:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ab9780d09290e29f7036a1ee8304360b30f2357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:41:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7a63b71d9c44541aa12cc4f75ae395c : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5afd31999314f40bd5bc2c6cb59d6c1 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2dc3a950b7f4e83833ee9478f6af544 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b5b7897de649d28ed9d1da8077e8df : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f142a96209214713b0028ec5ad9bc6b7 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5129f6652f1f4304b73f72c9c97d551e : RETRY_LIMIT in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65392aad27d468798655b57ea45c5ba : SUCCESS in 8m 58s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cff081ba0f644dad8365fda94133758c : SUCCESS in 35m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c41ec5d847f0452d93aa862562cf141d : SUCCESS in 36m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab21ba4c74514fa68393d523547c25a4 : FAILURE in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7d5b609a8689fb3edd55f0d407148084c87d11b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 09:48:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95b1c769385605304de40fd26f810c8eb26f3691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 10:36:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f75ca0e576e47aca7f5f811855da15c : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143cfdc14c6f473586c36f1bdb86fae0 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c61edc4a344679bdd50de2fabec399 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e897bef884994c7886611f24550f5319 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23ef991e8624413d9c46ddb87548b426 : TIMED_OUT in 42m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b27716cc2434ff38fa428a5ccdcf18b : RETRY_LIMIT in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c24f850c0a44d3be4bf8a87abf37ed : SUCCESS in 8m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f06a2ed999e94e9e82a13d5515e8e288 : SUCCESS in 40m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/48ecab158ab84729840f522ff08c9b18 : SUCCESS in 42m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af44c3ae20f04a7f93f373ff75fde8da : FAILURE in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"624f5093ebcc71bebed1a2cf7ec68ed486414a97","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-13 07:07:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5c39a5bc368b892483932ff7f1705d65cee40f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 08:04:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162363f187ba4a819a050a1496ce71b4 : TIMED_OUT in 51m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c25bbaa4b6c34b2bb45faa487e79f133 : TIMED_OUT in 41m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73a7f169b154238a77d27dccd19ff47 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/501144495f7d4afcad8a01f7dc43c622 : TIMED_OUT in 41m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78bbd5fac57b4c9ea6f3dbac107f8d83 : TIMED_OUT in 41m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29f1a585ea0b44a680b186b17c29a474 : RETRY_LIMIT in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5a733bf93b4b34bb68bc69096ad0f5 : SUCCESS in 6m 56s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4b3905d2fe42493fb9e727560d2d5f0a : FAILURE in 12m 17s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c9446be1578c4199bc322bf9c6c8f94d : FAILURE in 16m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3b6eeaa1e1241ea8d583391e5910aa5 : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d40b4b7012c11a413118f64ccaa58576d6b719a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-15 09:52:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9182b72dff3e9a1a0963e132465dd76907488387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:29:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0274c724f3c45a88ea473365748fb90 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8db28c60fb04d5489a5e058b8674ee3 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03efb564a054b85959afb49dfe93059 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/725a51d46d07489c81e28b31f84d6c49 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c380618ae58743d0a16b84a3e8d2daf0 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b54e130039f74443855e1b304473520f : RETRY_LIMIT in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37170afbaaa8468c8a6fb8bdb6bc3d7b : SUCCESS in 7m 14s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e76779e4f8d64a3693f196eaec4e535d : SUCCESS in 33m 04s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c3ffd01829ab4e2c830da6d1eaceaad0 : SUCCESS in 34m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12a00370a3544d2dac984505af8c46c8 : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1fc214de6dd7e2bf6054101a5659910446d99d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-06 07:44:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f26bb798a03b0f1f4e43c07a47ce5673624d9ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:44:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00e71ab16d4cffd683e361c91ee6ffea40c19a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 08:34:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e53c32e06d430bab1349522df67e82 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0aee53f4bd14f1f8462fa6d28aeeb6b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c0ddd1cd69475b8a817f315071fefb : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a38ed2e69ce4475b8d8b8c464acff4c : SUCCESS in 6m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/9c9bfb90046b48dc84671c670b177dfa : SUCCESS in 45m 51s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/d55858f681dd42ccbd78f57cbe6e7cda : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6869d9c7d8ac20203884c991c8240c34f7675a2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 08:34:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c27b19369f80e3677eee51552a87ec8f1f983504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 08:35:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58809ad4c97947f19a6cf69ef50a829d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd6469d53c692effdb04f67ce625b29f2d862d35","revisions":{"b62f70126b873575e4a40d512fb2df44993cab07":{"kind":"REWORK","_number":1,"created":"2022-03-10 15:48:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/833113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/13/833113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/13/833113/1"}}},"commit":{"parents":[{"commit":"730be45841e755ea672d10f4ac30f0594309556d","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/730be45841e755ea672d10f4ac30f0594309556d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-10 15:47:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-10 15:47:21.000000000","tz":60},"subject":"Fix references to start and end dates","message":"Fix references to start and end dates\n\nThe lease attributes are called start_date and end_date, not start_time\nand end_time.\n\nChange-Id: I77fb08a3811ed1445e31b3552ca23d5bfc43a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b62f70126b873575e4a40d512fb2df44993cab07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b62f70126b873575e4a40d512fb2df44993cab07"}]},"branch":"refs/heads/master"},"bd6469d53c692effdb04f67ce625b29f2d862d35":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-15 09:52:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/833113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/13/833113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/13/833113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/13/833113/2"}}},"commit":{"parents":[{"commit":"aa2a14a037da20571bd03b1dd1fa34566e0b1dca","subject":"Switch to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/aa2a14a037da20571bd03b1dd1fa34566e0b1dca"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-10 15:47:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-15 09:52:50.000000000","tz":0},"subject":"Fix references to start and end dates","message":"Fix references to start and end dates\n\nThe lease attributes are called start_date and end_date, not start_time\nand end_time.\n\nChange-Id: I77fb08a3811ed1445e31b3552ca23d5bfc43a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd6469d53c692effdb04f67ce625b29f2d862d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd6469d53c692effdb04f67ce625b29f2d862d35"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
