)]}'
{"id":"openstack%2Fblazar~835153","triplet_id":"openstack%2Fblazar~master~Ic55dba0038a2209e9fb291a604a5c1da57607e1e","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-15 10:10:12.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ic55dba0038a2209e9fb291a604a5c1da57607e1e","subject":"Skip blazar install in docs and pep8 jobs","status":"MERGED","created":"2022-03-24 22:09:50.000000000","updated":"2022-04-15 13:19:28.000000000","submitted":"2022-04-15 13:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835153","meta_rev_id":"139aebcf21832ebd8c394c59a83b39c7d43bf303","_number":835153,"virtual_id_number":835153,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-15 13:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-15 12:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-15 12:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-15 12:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 22:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ccccf211b13c0ae9f88fa26c09cc2c263f19ce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-24 22:09:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca57cbb00d0131460a0a968f36de869f9040503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 22:56:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/087fb65040f04c7a8e46781d310a52c0 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/890eb8ca4c3a4e82bcfb3d2e60357136 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a347d220a3034da794fed016986b3bef : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71833cfbabf04c2fa25c3389008e5475 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab8389c502c4b89a548115a45a6e65f : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e80679685dab421bbba5cdc0af2ddb74 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/194118d96b6c48f6acddd18fe89d8927 : TIMED_OUT in 42m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80392bbdcbc0450b85f2fa55d5e54698 : RETRY_LIMIT in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e47f0cbfcf4394a3b2455c699b77ed : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e03a6bed54f4925b82be20a6e905b45 : SUCCESS in 5m 59s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/da44d637a705447fbea6f1ea6b7a13e1 : SUCCESS in 37m 14s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/83fa6c43f9d84294af168f42b1d3081a : SUCCESS in 38m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0eb13af4b17f46fbb76285235253d206 : FAILURE in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0af446c865ab63cb111ab35f00cdafbbcc801d42","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 08:33:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03094e4f7bbac76b6250619f646273e9a12290b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 09:25:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c37c88be501f4b158c3f435c8ee32667 : SUCCESS in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e739c6b29abd452b9a11270a96ce5baf : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17fa7a44e6a4e9fbc84b8a12a7fd42c : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17ebdb936c744e0aae31b7837345c80d : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7647bfc9e0b4336ae2d533dfc41f57c : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a610da99c6f7428f9a9fa22be8a7bc65 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eccf15190264f04b685bb7716910d91 : TIMED_OUT in 42m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21eb94f2d76b4919acb13a88a0fd0ded : RETRY_LIMIT in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4212c032419248f8800c77941f0023ee : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0f9ea4293c6445da874448d5e29b22a : SUCCESS in 5m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2a15270f9d324c4ca2452b866496ea42 : SUCCESS in 45m 53s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/33db0786263f475fba02f8b67172069e : SUCCESS in 44m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/344276881a51437baa72cba23692e5b2 : FAILURE in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"666d87a1be64d50e47b182419e856e9e62ec089c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-15 10:10:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d3126704b157ec12cdf360d86fce8f702bc6dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9055fb6a3edd454287992fd7e6b66a66 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf54580a20ca4c6cb2a274e24ec452c7 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b65c5e670d724b9cb3d5a35e0e839ad4 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37e36a92c0274f949d744d5eecd270b9 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64bbc0ec273466da9c9728dd5e96edc : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e387413c56534e4fb6678b1a724ca959 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d75a846de3348629b3dd3329c0b4dfc : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f703081198d442d91a693c6cf0c0855 : RETRY_LIMIT in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcded36ba7214ff38a146e5c477e5ead : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6160b9280c14491a49c20b032e47567 : SUCCESS in 4m 43s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/611db313f8254982bda1595e329187bf : SUCCESS in 32m 55s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/557ad9f1129340588c01a43c359852c0 : SUCCESS in 47m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81083d5719cf48d4b5f1f7e8ecc598ae : FAILURE in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b4238d85fedcf48cb4d8ed33abb8ee9b19bfeb4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-15 12:28:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c53d2a3aba31f27089679eabd4e9daaaf4fbff21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 12:29:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5064f16370271ead57b7467ba8f6970577881d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:17:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7b9ac056e22648d7ac73522b0923b27d : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/46925e78edf64147b855af953068a2a2 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42d925ebc6fc4751b8e7e5ab85f02ada : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1fb4a555614ad69db5f7b51f8bf8b5 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb0970d58c2140858357965101542189 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9856211baa64f0bb6d5cca042c78ce0 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d634c40f6e904142827412fb5df0bdc7 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54d7bf6ec38d4ecdafe94cfa6bb76f34 : SUCCESS in 4m 09s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6779b5737b77461db621d1ae1a72ac8a : SUCCESS in 44m 32s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/fab45572e6d849d5bea2fa8a2d8b1423 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8f5ca1ad911c3f0ac48ffc008fd26de99c1e3d00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:17:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"139aebcf21832ebd8c394c59a83b39c7d43bf303","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:19:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e9713d1e5c9841e5a73b8bdeb7fbe6eb : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6baa93ab10e4486f922736b925370dd3 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e76372ff4a2f422282f2804da88c17dd : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c3d94ce3ba74b0d66772ae887567894c930af2b","revisions":{"2c3d94ce3ba74b0d66772ae887567894c930af2b":{"kind":"REWORK","_number":1,"created":"2022-03-24 22:09:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/835153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/53/835153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/53/835153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/53/835153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/53/835153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/53/835153/1"}}},"commit":{"parents":[{"commit":"730be45841e755ea672d10f4ac30f0594309556d","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/730be45841e755ea672d10f4ac30f0594309556d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-24 22:09:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-24 22:09:24.000000000","tz":60},"subject":"Skip blazar install in docs and pep8 jobs","message":"Skip blazar install in docs and pep8 jobs\n\nAfter installing doc build requirements with upper constraints, tox was\ninstalling blazar without constraints, which was pulling Jinja2 3.1.0\n(released today) as a Flask dependency. This caused a docs build failure\non stable/wallaby, because Jinja2 dropped support for environmentfilter\nused by the version of Sphinx in upper constraints.\n\nUse skip_install to instruct tox not to install blazar when not needed.\nThis requires adding oslo.config and oslo.policy to doc/requirements.txt\nbecause we use their sphinx extensions.\n\nChange-Id: Ic55dba0038a2209e9fb291a604a5c1da57607e1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2c3d94ce3ba74b0d66772ae887567894c930af2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2c3d94ce3ba74b0d66772ae887567894c930af2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
