)]}'
{"id":"openstack%2Fblazar~840035","triplet_id":"openstack%2Fblazar~master~I24fd4700d59b30e4e5b3fed6c892a4f8ade7bb47","project":"openstack/blazar","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-02 11:55:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24fd4700d59b30e4e5b3fed6c892a4f8ade7bb47","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-04-30 20:58:06.000000000","updated":"2022-05-02 11:58:00.000000000","submitted":"2022-05-02 11:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840035-drop-l-c","meta_rev_id":"a05b5da45733d999d246d50ddf4228a99c76eb5f","_number":840035,"virtual_id_number":840035,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 11:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-02 11:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-02 11:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-30 21:44:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 09:23:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-05-02 11:15:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"1a7ab10970f8aa292273809d5edfe634c347342a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-30 20:58:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"417eb4055b6b79c2315d463cab8b82ba10adc81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-30 21:44:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ed0738ea0a294bfcaab625bfc798b938 : RETRY_LIMIT in 1m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14167093386e43a789384b1890c06ae2 : RETRY_LIMIT in 1m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e8a8843bd68454ea72e43f3750c135b : RETRY_LIMIT in 1m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a401899ca75c4e21bf3b53af383d5928 : RETRY_LIMIT in 1m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b014b0c087d245f1bc85e040f02aa813 : RETRY_LIMIT in 1m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/163f3f0c1a244fd68db586ddfe6a347e : RETRY_LIMIT in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db1a9baaf0c948b68aeed5fe9ac042d7 : RETRY_LIMIT in 1m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120fbea96f0745ea9425c345aa7ac2b3 : RETRY_LIMIT in 1m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bb4369132434e6f8d565e80fe176a0e : RETRY_LIMIT in 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1381312c495c4bb5b28e3d034f5348bc : RETRY_LIMIT in 1m 19s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5367c8e6fc374e33b5fe84d24212a1de : RETRY_LIMIT in 1m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86aa1b0dea174e6d946e6593f3d3d634 : RETRY_LIMIT in 1m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b921bef560af4f04a95349b6767ad558 : RETRY_LIMIT in 1m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2f47165d490d983b05d87dba2ccffcd13865356","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 16:55:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24b15717b529b578dcd4b7df6bf9359429c43be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 17:44:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6f00b70781a14e4cba3956a1386ad4db : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca5010a936c6458cbea93e56733242de : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8d829ef435c41fbb67ad5f5d37a9cdb : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e995f6f4924c9a98ad79a464dc7171 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50436f8df7aa43248178f3a5c3aab7e7 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74291ec1f6694e9d91d847bd3b9f71fc : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc8bf5d380214a06837823cae7299fa6 : RETRY_LIMIT in 2m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc4d6c2473f4007b7707eb46a3aa429 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76a2d55af2f24c75bf439b7352af4ddb : SUCCESS in 3m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8ff045ccbbdb443aa4d0f7da2d1b24d7 : SUCCESS in 40m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0e3516f3772f421db38a1ea1e39409ee : FAILURE in 42m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/746f9fa0e5044098bd7a55dcc0551464 : FAILURE in 3m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0abc749cb1d54488b2b907922bb2f9ef : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5afb00c78787b6d4e5172cf201768b32ac2673b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 09:23:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3401c10a95862e6c2174ffcf8f92b1ffed04bb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 10:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2459db609a344f15aab2025e29f84c21 : SUCCESS in 4m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f458f7e7031e462fb3387db24b66f607 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09de9dd2539e41de9d2bc976aea4e2e2 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61e3384608247b49df82239f0780e29 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf829f193cc247b785698cdfadf7e538 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac780e4beeca4ad999e8e8c6db74ebd6 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0a2e95b481546319bd7dd0eb0fb2627 : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5cbac8edb334027ab2ef1e56537936d : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4ec033c83cb4c8b8bd922456d36c072 : SUCCESS in 4m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/34072f679c1348e0a67feb16eab94b96 : SUCCESS in 37m 45s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a70ef68d12284aa2bda4c2c235e25c90 : SUCCESS in 44m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a0ae49eb29f4436930ba87700e400e1 : FAILURE in 3m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b199c1f61734783bcd8f459495a8b03 : FAILURE in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a1d24fb0a62475ae3d2dd04f8d099317828ab6d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 11:15:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f97c6da32e226478f18e8bfc39a3b0aaa3baf05b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:16:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"262b4761403425298a9c8c60bfbe159a355701ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:55:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6718811f00ab4ca8af06cd07e9478b73 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f90c00bd7ee49dfba38c9a106dcf357 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df08bbbd5027465d88dbb5297f8947ca : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f95ac6601c554e0a84c4944a1b187664 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7c66741862f416b8b38ca62a0377c8c : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4102365ca04fcab4973da34b24fb38 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794da89417b04b7e99d8fa3b8c8084d4 : SUCCESS in 4m 12s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/38cd1eb1df754faba4dcecc3771dfceb : SUCCESS in 37m 37s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/654f1601290c48bebd9287e55745cedc : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"25e262a36f5824469f1f29a2f39c3b360b291cf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:55:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a05b5da45733d999d246d50ddf4228a99c76eb5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:58:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6028d4bba40948669e5359dc557a37dd : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d3507e49924347b9fa0cd445a7e9cc : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ca325cb3120448490e3212aee3d4445 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d47b5347b60f6419aeacdeffe10642ffc1d9362","revisions":{"7d47b5347b60f6419aeacdeffe10642ffc1d9362":{"kind":"REWORK","_number":1,"created":"2022-04-30 20:58:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/840035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/35/840035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/35/840035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/35/840035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/35/840035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/35/840035/1"}}},"commit":{"parents":[{"commit":"66291b6e1071f8fa7c8080ff3c9fa297921d0c4e","subject":"Merge \"Skip blazar install in docs and pep8 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/66291b6e1071f8fa7c8080ff3c9fa297921d0c4e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 20:32:04.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 20:32:04.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I24fd4700d59b30e4e5b3fed6c892a4f8ade7bb47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7d47b5347b60f6419aeacdeffe10642ffc1d9362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/7d47b5347b60f6419aeacdeffe10642ffc1d9362"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
