)]}'
{"id":"openstack%2Fblazar~851973","triplet_id":"openstack%2Fblazar~master~I28d06ae31c4cf44d010e0892a094138112b08641","project":"openstack/blazar","branch":"master","topic":"fix-docs","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-08-23 12:57:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28d06ae31c4cf44d010e0892a094138112b08641","subject":"[docs] Install blazar to generate proper docs","status":"MERGED","created":"2022-08-03 09:24:11.000000000","updated":"2022-08-23 12:58:49.000000000","submitted":"2022-08-23 12:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851973-fix-docs","meta_rev_id":"b3787f10653851f10715160f139561e14e2ef2c0","_number":851973,"virtual_id_number":851973,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 12:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-23 12:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-23 12:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 10:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 12:02:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"74cbcb3d3dd70deb02a1eab440b4a8ff90dee6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 09:24:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07c59f3ac8b1235c43679c755b936d91e39ef019","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 09:25:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"679fb5adf78938f462faaef7d877392f3a9c9c4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 09:29:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9635265b5552888e792a49e591360fb10290a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 10:00:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2334c166b3a14090b6be3529bb071d53\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00c03551e1e4411789355ed1c8fe83de : SUCCESS in 4m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/451d17251e244f8f8bf11ef6ce9252b7 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb5b37c1ecd44adbcf3823746ff7e43 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57e843812d14b5aa88a1a0623c58ff0 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52bc13251e74e9ea03e92a5ab48f8eb : FAILURE in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d75f9ec47d44b2c86dbd2ba32e394e2 : FAILURE in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fccf9838252d4de5b3b956bd34ce7326 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2eebc4c98444bf0acee094a3cc46683 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e275df640a6946d0a723b9de7e83cd39 : SUCCESS in 5m 26s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f6aaf8a9b4104e99a0e2df451f472022 : SUCCESS in 35m 55s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e27edc1dbc4b4417a46d3506f7b75934 : SUCCESS in 34m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aeb2d58151304d7abc48e3c1581316ec : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdd7422f7d4028b4cafb2aa24238f8616b34e5fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8bacb27fb63b4fba1b8a262ef8dcf8f21109627","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:05:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ac4e11ec0a5e55d56dad287470592ade015f464","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:16:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7decbf8536ea17871fada4090bcb64a98f722d55","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:17:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d73abee3306087316a706686c401a9e11630b266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:17:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d003e66ce116e4c19c05f5e76b5744772bced236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-03 10:40:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5ce9d9cc603e1f5a6758c4ca96c777c0effea554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:27:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9e9a58a1b8d49c2ab4eb135f64284ab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9d4666dc13b643f381440ecd87f1177c : SUCCESS in 5m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd3216fdc9e141fab73f3dc75b3e39d2 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b338b7fcba704db0aaab891c161f8319 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7517d06a9243da8ad450dc3373f3bf : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c21d020604e8420eae18abb0db8c5e61 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5884c18af2bc4d718eab4b384a5b56b7 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48cb97dbbe904c20b05ba770cd881209 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3796718383e140888739dd3036681595 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea70cb9f71f24a6890c877f62e8df3a2 : SUCCESS in 4m 23s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/71a8e7e8bb9643ebb45d3dfa82a01d55 : SUCCESS in 34m 25s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a6affb4661da43d8b3e00d3dc52701c6 : SUCCESS in 45m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/871e70fb336b4506861fc889bf0281cd : FAILURE in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42825edc8c05c318cba1e839f888324203e34cc7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-23 12:02:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d25d71f6262cd62093115d685b69d762db4fe40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:03:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b45d3d50b115ab626eb6075c85d3c9eb5f87715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:57:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54bd738008145b0a4b89ba893cfe4c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab75d3e8d6d342359c3c27b283b276d1 : SUCCESS in 4m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6033104ea61543a39b82f35af39c4ca5 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c1e7c6de204cddbbb6a4ed5befeea4 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b1c699c42c4ff598d5f39948664cd5 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec15a91fd4a74a7ea7f442bf22545f72 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11be96d663f9423caafc31d53cee2eb8 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c54f72c69bda4d86b0ccaed2abe17a75 : SUCCESS in 4m 23s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/06df0432750841c88e33ed18a42952e9 : SUCCESS in 49m 36s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/209d43f4580f4f3ca2d5f4b18ae2a165 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a55098ff7554e8fff589171b66e983c36eb01c36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:57:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b3787f10653851f10715160f139561e14e2ef2c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:58:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff6ce140db641ee8bc1cf92e498bb47\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0122ce52ee67420ea205d13c67fe98cd : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22627320c0a4473b5e665108aab61ba : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/057ddd10bc034b5b914b8c4639695262 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"42e4b7639f4673867e6868e13669c83ca46dcc3b","revisions":{"d27a4ad20797429f71fc67511da022c36ba339f6":{"kind":"REWORK","_number":1,"created":"2022-08-03 09:24:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/73/851973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/851973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/851973/1"}}},"commit":{"parents":[{"commit":"bda5168336477e4d97abf6191b3c7162b4144191","subject":"Merge \"Update python testing as per zed cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bda5168336477e4d97abf6191b3c7162b4144191"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 09:20:26.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 09:24:09.000000000","tz":120},"subject":"[docs] Install blazar to generate proper docs","message":"[docs] Install blazar to generate proper docs\n\nThe blazar installation was removed in [1].\nHowever, stevedore requires blazar to generate docs for\nconfiguration. Currently, they are mostly empty.\nInstead of removing the blazar installation, this change makes\nsure that deps are installed with constraints so that blazar\ninstallation does not bring havoc and the original issue is fixed.\n\n[1] Ic55dba0038a2209e9fb291a604a5c1da57607e1e\n\nChange-Id: I28d06ae31c4cf44d010e0892a094138112b08641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d27a4ad20797429f71fc67511da022c36ba339f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d27a4ad20797429f71fc67511da022c36ba339f6"}]},"branch":"refs/heads/master"},"d549cd15cbfc0c86f47d8322f86e1c38eaba7c39":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-03 10:17:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/73/851973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/851973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/851973/2"}}},"commit":{"parents":[{"commit":"1f1ace3003c2e31296133d30f0e323197790eefa","subject":"Add WebTest as a test dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1f1ace3003c2e31296133d30f0e323197790eefa"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 09:20:26.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 10:17:54.000000000","tz":0},"subject":"[docs] Install blazar to generate proper docs","message":"[docs] Install blazar to generate proper docs\n\nThe blazar installation was removed in [1].\nHowever, stevedore requires blazar to generate docs for\nconfiguration. Currently, they are mostly empty.\nInstead of removing the blazar installation, this change makes\nsure that deps are installed with constraints so that blazar\ninstallation does not bring havoc and the original issue is fixed.\n\n[1] Ic55dba0038a2209e9fb291a604a5c1da57607e1e\n\nChange-Id: I28d06ae31c4cf44d010e0892a094138112b08641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d549cd15cbfc0c86f47d8322f86e1c38eaba7c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d549cd15cbfc0c86f47d8322f86e1c38eaba7c39"}]},"branch":"refs/heads/master","description":"Rebase"},"42e4b7639f4673867e6868e13669c83ca46dcc3b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-03 10:40:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/73/851973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/851973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/851973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/851973/3"}}},"commit":{"parents":[{"commit":"d02152fd9783a9bd34f72357a57ed4d6ab00d4c2","subject":"Add WebTest as a test dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d02152fd9783a9bd34f72357a57ed4d6ab00d4c2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 09:20:26.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-03 10:40:57.000000000","tz":0},"subject":"[docs] Install blazar to generate proper docs","message":"[docs] Install blazar to generate proper docs\n\nThe blazar installation was removed in [1].\nHowever, stevedore requires blazar to generate docs for\nconfiguration. Currently, they are mostly empty.\nInstead of removing the blazar installation, this change makes\nsure that deps are installed with constraints so that blazar\ninstallation does not bring havoc and the original issue is fixed.\n\n[1] Ic55dba0038a2209e9fb291a604a5c1da57607e1e\n\nChange-Id: I28d06ae31c4cf44d010e0892a094138112b08641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42e4b7639f4673867e6868e13669c83ca46dcc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42e4b7639f4673867e6868e13669c83ca46dcc3b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
