)]}'
{"id":"openstack%2Fblazar~862799","triplet_id":"openstack%2Fblazar~master~Ibed073bf14ce5ea23641aa8a1362f239351aa80e","project":"openstack/blazar","branch":"master","topic":"migrate-to-jammy","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-11-25 17:28:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibed073bf14ce5ea23641aa8a1362f239351aa80e","subject":"DNM: testing jobs on Ubuntu Jammy (22.04)","status":"ABANDONED","created":"2022-10-27 11:48:56.000000000","updated":"2022-11-25 17:28:23.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd2aa6c69d618b7dff00682e74c5af75a16bb529","_number":862799,"virtual_id_number":862799,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-17 20:56:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 12:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 09:55:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"6353b10d2c289fb6461852d47cacb3dc39ffeeeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-27 11:48:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"493b572d4414898eaa9241fb26008b8c2d72dc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:37:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c309b2bf46d24764bd4b97af8542fd78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f4347a6d4454fd8a8076b5a242af15a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bc65fc24a843608c5377a6a3e1f4f2 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f9e2c4fa854eb1a90f7968e8d75ea2 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f986a5d36e846d28173d6d571fece45 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1121309e004042d0acc1e1be49a153af : SUCCESS in 7m 12s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4c8d99dfe9da4008b6ebe6ddaf1dd774 : FAILURE in 40m 26s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b3e8fc85542e47f38fb3c71ad940c84f : FAILURE in 40m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ede1369fe9e4cde9f397fc2e7b23d58 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ec08d0c5373076051c0ee399d0e23b9d3961c45","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-16 09:55:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5240104e42b64125c03f8814dc0a28ac229e2ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 10:44:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95c3fdb412fe41199ce18d577dad2f5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e6f85427a3948bfa8a2d4fe51b75ac9 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b217b3a1434ba1be4a85097e0161f0 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60b1f7f7b8f74196a2cf9f12ac311385 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e2b5f0bb650467ebbab881822fb4bc0 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e51c90bebe49d4ad5463750bcb9f9a : SUCCESS in 7m 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c3d080d66ecb40bc80ff50d4bef192fc : FAILURE in 41m 10s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/373f484cd6344dedba71ff1b1247a4e3 : FAILURE in 33m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8803fc1830349639bf3d71c44f9e15f : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e469a1a07e08efa97702af3b816b5db1912b50f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-17 13:18:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bed163b2c145d1f0a3c20c4479b7adb188153c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 14:00:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c98c168b7cf43089dbaf30a17a4ac2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70961195330d431a9dc301f5aa476be3 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aac7df4c61c4e779baad6cb1734fd3b : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/958688e68cc345fa85dfdacfe81669c6 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/684ca77070994e96a2de064d3a71e59c : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678e03897dfc4617b0b5b7f223304b27 : SUCCESS in 7m 16s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/84775c526ce244dab8ca5d29b90b9612 : SUCCESS in 33m 00s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/98578b3536a6438099ceebedc01880ed : SUCCESS in 34m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98afdad1054b4ba2ba6625991e7a558c : FAILURE in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32769f5a9474d21619982786fbe13143fd14a16c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-17 20:06:33.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"04731a71bf8f1c767770518949f01a351ce91cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 20:56:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f1473ca32e24c208b52d013f8534b55\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e06446893c464a97dc37feaf3286ca : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee81fbb752124c1db99ca17ac7ce9c15 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e73959271a64e1e8d9a23a53187d539 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06fc32ceaac04a269d32e3539dc120c2 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd284273f42444b815daf0df88dc440 : SUCCESS in 7m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a4630d6c47a64252b3020123a086fe1f : SUCCESS in 44m 03s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c3effd8038614071a0dc2e768d5e008f : SUCCESS in 44m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8868665fdaeb4ba5900ba614d0cc618b : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd2aa6c69d618b7dff00682e74c5af75a16bb529","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-25 17:28:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2658339a0dd32dc5e5d69035ff009d5f4df244e","revisions":{"80aa3c4fabcdec646eef246029568bc6361a0383":{"kind":"REWORK","_number":1,"created":"2022-10-27 11:48:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/862799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/99/862799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/99/862799/1"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-27 11:48:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-27 11:48:54.000000000","tz":120},"subject":"DNM: testing jobs on Ubuntu Jammy (22.04)","message":"DNM: testing jobs on Ubuntu Jammy (22.04)\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/861110\nChange-Id: Ibed073bf14ce5ea23641aa8a1362f239351aa80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/80aa3c4fabcdec646eef246029568bc6361a0383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/80aa3c4fabcdec646eef246029568bc6361a0383"}]},"branch":"refs/heads/master"},"ab4cb579c17d5dff5e4f5134f2abca9c35e11a97":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-17 13:18:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/862799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/99/862799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/99/862799/2"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-27 11:48:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 13:18:23.000000000","tz":0},"subject":"DNM: testing jobs on Ubuntu Jammy (22.04)","message":"DNM: testing jobs on Ubuntu Jammy (22.04)\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/861110\nDepends-On: https://review.opendev.org/c/openstack/blazar-tempest-plugin/+/864895\nChange-Id: Ibed073bf14ce5ea23641aa8a1362f239351aa80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ab4cb579c17d5dff5e4f5134f2abca9c35e11a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ab4cb579c17d5dff5e4f5134f2abca9c35e11a97"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a2658339a0dd32dc5e5d69035ff009d5f4df244e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-17 20:06:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/862799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/99/862799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/99/862799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/99/862799/3"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-27 11:48:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 20:06:33.000000000","tz":0},"subject":"DNM: testing jobs on Ubuntu Jammy (22.04)","message":"DNM: testing jobs on Ubuntu Jammy (22.04)\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/861110\nChange-Id: Ibed073bf14ce5ea23641aa8a1362f239351aa80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a2658339a0dd32dc5e5d69035ff009d5f4df244e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a2658339a0dd32dc5e5d69035ff009d5f4df244e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
