)]}'
{"id":"openstack%2Fblazar~874873","triplet_id":"openstack%2Fblazar~master~Id46d5c8ad3695c1a0f899350aac8469636f3b5b3","project":"openstack/blazar","branch":"master","topic":"release-health-check-cwr-antelope","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-24 10:41:57.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Id46d5c8ad3695c1a0f899350aac8469636f3b5b3","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-02-23 10:43:32.000000000","updated":"2023-02-24 15:49:46.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3c08986bb5b975276d8420b8db738e544651bd8","_number":874873,"virtual_id_number":874873,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-24 11:15:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 11:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"654882ebe8432e7a02107c3a8bf09050e69a8fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-23 10:43:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9a7d056301d640472201feb297a9591f0c47c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 11:41:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08dbaae5ccce49cf9a2bc0ce41ebed61\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ca3c94d9edf430eaa79d7b23fa43a45 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9906bf56a5645c68e73c8adf5561828 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/089aa92815a548208e320d4a5da070bc : FAILURE in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36f59c2ec0904fb3931d887b2ca50a9e : FAILURE in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82bdf95140554e6192f22fc0789006a3 : SUCCESS in 6m 52s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/fe8488a6674f42228d3922d1fb41dc64 : SUCCESS in 45m 21s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/39decee5c0a0459b90ebe59701b496a4 : SUCCESS in 46m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e41492c1f1f7438da94d5de0c0015770 : FAILURE in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c41589018514017d3e23bc5b5ebc751c90a28a8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-24 10:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9c66a664d80a8f8f81da64f57581ae4d6ac8d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a635930a20476181f93fd714440580\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4adef1d0ab04a808e23d16d7fe75450 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed799f97993d4a5085ff8ae2774fbc99 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b16cfe91c2794b6fb787960234eedddc : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad2645ca136c47c588f114825cca1dc6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3ea4a5f26649bb824d5ece662b63ca : SUCCESS in 6m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c33359c5baff441885b2b1547f5e64a3 : SUCCESS in 30m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/95134292dfb748a5b24934dd1d0c605f : SUCCESS in 31m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b9e9a7aa08f74823a4af988e4c6be648 : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3c08986bb5b975276d8420b8db738e544651bd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-24 15:49:46.000000000","message":"Abandoned\n\nGate is unblocked, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7776abe2d43a2d071885accfe9a4bf28b8dc526","revisions":{"a7776abe2d43a2d071885accfe9a4bf28b8dc526":{"kind":"REWORK","_number":1,"created":"2023-02-23 10:43:32.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/73/874873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/874873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/874873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/874873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/874873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/874873/1"}}},"commit":{"parents":[{"commit":"68c6487465d15bf9db4559e36eaf2b50ab2a0a9c","subject":"Merge \"[docs] Rewrite the architecture description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/68c6487465d15bf9db4559e36eaf2b50ab2a0a9c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-23 10:43:23.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-23 10:43:30.000000000","tz":60},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Id46d5c8ad3695c1a0f899350aac8469636f3b5b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a7776abe2d43a2d071885accfe9a4bf28b8dc526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a7776abe2d43a2d071885accfe9a4bf28b8dc526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
