)]}'
{"id":"openstack%2Fblazar~896304","triplet_id":"openstack%2Fblazar~master~I12009bdef08a435cdafe604f0832a8dcbdf16b24","project":"openstack/blazar","branch":"master","topic":"bindep-bookworm","attention_set":{},"removed_from_attention_set":{"35978":{"account":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"last_update":"2023-09-29 09:59:35.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-29 09:05:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I12009bdef08a435cdafe604f0832a8dcbdf16b24","subject":"Update bindep package list for Debian Bookworm","status":"MERGED","created":"2023-09-23 06:48:44.000000000","updated":"2023-09-29 10:00:35.000000000","submitted":"2023-09-29 09:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896304-bindep-bookworm","meta_rev_id":"dfebf4bdd367424576e17b928a286268518e777f","_number":896304,"virtual_id_number":896304,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 09:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-29 09:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-29 09:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-23 07:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 10:17:59.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"REVIEWER"}],"messages":[{"id":"eaa354d1509dfcfd070e1201b8ecdeb37c628edd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-23 06:48:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a8688f0fca9e5a47e7472313ab7a6bbd108af4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 07:41:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f65da54fa3842afa46ae4e0a9a3e0b7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bd737d3b58f444fa895f2fda7723510 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4921dc55dc71439cbbc1b7f4b411c84a : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a295dc4a78624eb9a41d5389f6fd2de6 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9be3ddccb45e471d8aa045c90245632c : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35522a292ba46c5a50ed37bb16e1653 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53b2d5e973bd407c8de1b2d9d923eb60 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b54a002399be415483114cb5120692ff : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a231ae22164b209680125deb968f79 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8370fe82154649fca487f9f7e85652e2 : SUCCESS in 3m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8489560469c04b56b52d6c96ae62ba6d : SUCCESS in 41m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a719667d7023450a85b31259a1eb93f8 : SUCCESS in 45m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d764a0001d654dda962b089b41cfacf0 : FAILURE in 3m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa6c9055e165ca054439f3357ac3246ca22b498f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-25 07:46:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"db0abaf575448bd6f31aa098daa88c565965acbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 08:39:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896ba9b52b0e4da885cac0b4c75e11b1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bd4833d2f46b4ef4892e2508684fb9bc : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6698e341e718482ca3d741290a637b11 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6201104e711f4161926a95d71fd2780e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7110ecd69fb54e84828d795292816277 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8537a71572374052ae2cc5b0e21ceae1 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4803e8d12cf4b069487a214c85db8bb : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44a9699d5c9549fe9f4767d5c9cb5df6 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c127ce22c836476d9d1368d80401619d : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec9d7a371ddd4b998caa98a66895ce93 : SUCCESS in 3m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6be31647b6b246b6a5d4528fd43eaa9d : SUCCESS in 28m 33s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/aed3437163b540069ba6a494e64e2b9e : SUCCESS in 45m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/590f7287d2d34475ad602422507e277b : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"383dac394ff656da09fbcf46b286279e3f92f3ab","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-25 10:17:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5036ced3b557aee144a4f335eccd2f87dec074ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-29 07:47:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"38c79f6a9427881e9966f2b160c77eeb99a69263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 08:44:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d76131c8e6441dbdd2a289b14679a2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d49b9b8c24a842a585bcb6197da186f3 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ba5b9318e54c60aeb98b1152cf8cdc : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3d46354e6646b58cfc67be6e36e50d : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ee9b39335554fe3953989b2e001666f : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b812bbbc475e4ca6923293670193045d : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1045fcc6c95a42c2aff3fbd490cbbc45 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2bf7ae6c4b04fb98a5beda06cf1b861 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7185748c66bb4210ab8fb9f7108ab405 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b6a0aca3ff942fba23688e125490264 : SUCCESS in 3m 42s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2a1f1afd89ac4dcfa209964873c41a95 : SUCCESS in 44m 07s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f7eb8388a2414dda9367652cce3ef695 : SUCCESS in 51m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86082614a5104ad88a811008bfa8b812 : FAILURE in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e1318a4d80d9e96376b98c18bdc9f3efa239306","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-09-29 09:05:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"046cbc3262ee4bda5800138cdcc501d4e9a2df93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 09:05:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d6682c68c7fc3bd1683ca3700011e01da436658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 09:59:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20d2cbda00e4bce84a2f9172f961fec\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/958793e72f9748a2994161ec5c6276b2 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f73284c43d4093a33f4e1fea4d4118 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee57c49a5914428397cc9e7fe6a290c1 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b276478c3d14358a3cce020c80df807 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6592d3ee92674e98a8ce04cb67fd2148 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1103677ce7314c68ae04b73bc1f95f7e : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a400473dd584490857a6122387d5cdd : SUCCESS in 3m 49s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a07270f72fef4e918005fc5212ac6f0e : SUCCESS in 48m 34s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2da981ba1d4aa2a0e82232de36595b : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3cae717913071ecdab0b4a447c828acd11939d2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 09:59:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dfebf4bdd367424576e17b928a286268518e777f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:00:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a770d8a514849aa86612d7eb56d92a0\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e0b3017db8194b0faa3a658ee357ab99 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fb6a98a37444059c865ef52accef0f : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/022ca88a4c134a57aa2964a5a6ff60d9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"667a8bd782579a81fab0a5048a6d44fc2bfc7059","revisions":{"fec04d643b4b5262702f40ba7eabbe414c87458e":{"kind":"REWORK","_number":1,"created":"2023-09-23 06:48:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/896304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/896304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/896304/1"}}},"commit":{"parents":[{"commit":"e78848edb233751c78947282d120e82bc380a741","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e78848edb233751c78947282d120e82bc380a741"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-23 06:48:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-23 06:48:25.000000000","tz":120},"subject":"Update bindep package list for Debian Bookworm","message":"Update bindep package list for Debian Bookworm\n\nChange-Id: I12009bdef08a435cdafe604f0832a8dcbdf16b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fec04d643b4b5262702f40ba7eabbe414c87458e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fec04d643b4b5262702f40ba7eabbe414c87458e"}]},"branch":"refs/heads/master"},"12f2fcf5e6c59d93a293033583a048449b6d729d":{"kind":"REWORK","_number":2,"created":"2023-09-25 07:46:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/896304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/896304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/896304/2"}}},"commit":{"parents":[{"commit":"e78848edb233751c78947282d120e82bc380a741","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e78848edb233751c78947282d120e82bc380a741"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 07:46:13.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-25 07:46:13.000000000","tz":120},"subject":"Update bindep package list for Debian Bookworm","message":"Update bindep package list for Debian Bookworm\n\nChange-Id: I12009bdef08a435cdafe604f0832a8dcbdf16b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/12f2fcf5e6c59d93a293033583a048449b6d729d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/12f2fcf5e6c59d93a293033583a048449b6d729d"}]},"branch":"refs/heads/master"},"667a8bd782579a81fab0a5048a6d44fc2bfc7059":{"kind":"REWORK","_number":3,"created":"2023-09-29 07:47:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/896304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/896304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/896304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/896304/3"}}},"commit":{"parents":[{"commit":"e78848edb233751c78947282d120e82bc380a741","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e78848edb233751c78947282d120e82bc380a741"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 07:46:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-09-29 07:46:36.000000000","tz":120},"subject":"Update bindep package list for Debian Bookworm","message":"Update bindep package list for Debian Bookworm\n\nChange-Id: I12009bdef08a435cdafe604f0832a8dcbdf16b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/667a8bd782579a81fab0a5048a6d44fc2bfc7059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/667a8bd782579a81fab0a5048a6d44fc2bfc7059"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
