)]}'
{"id":"openstack%2Fblazar~905954","triplet_id":"openstack%2Fblazar~master~Ib070ffb0eb85ab0d5a299ca042a0646e8529fcac","project":"openstack/blazar","branch":"master","topic":"bump-eventlet","hashtags":[],"change_id":"Ib070ffb0eb85ab0d5a299ca042a0646e8529fcac","subject":"bump eventlet to latest version that support python 3.12","status":"ABANDONED","created":"2024-01-17 15:38:27.000000000","updated":"2024-01-19 13:01:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b68a351fe29dd89064151021c4da256c62cbd052","_number":905954,"virtual_id_number":905954,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-17 16:35:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 16:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23b81d31e28a721d7e067ffa79aa1a997b8410dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"244a54411e5a31a7169a7e8397fd8b01cdb2a395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:35:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6532b6b6214f4a0f8612ad0ad17f1f0f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c78e4f045c1b4295b9324cfecbc1703d : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ffe2572088d4a8293d72943198114e7 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9ae2bdcbb647a9bde46223cfc98c99 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d56bbe50936343d1b14134f5139cae1f : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4eea0db51fa64f93a5cab0700b996819 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350ce1abd1a04401aef2a606a3e9ea4c : SUCCESS in 9m 06s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cf84b6d7327c407cafcd39e1d66d3800 : SUCCESS in 46m 51s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/59b892493e5f4be3a7764eb1a43db369 : SUCCESS in 39m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/909b56fd07544eb9be94e58eb42f04bf : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b68a351fe29dd89064151021c4da256c62cbd052","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-19 13:01:32.000000000","message":"Abandoned\n\nThis version have been reverted from Upper Constraints. https://review.opendev.org/c/openstack/requirements/+/905905","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"daff5843bff2b8f56f1fa9ba9643502f05355fb4","revisions":{"daff5843bff2b8f56f1fa9ba9643502f05355fb4":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:38:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/54/905954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/54/905954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/54/905954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/54/905954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/54/905954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/54/905954/1"}}},"commit":{"parents":[{"commit":"b0dbd0a9299cab3451c1b3abdd35f6b5e76ca14a","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0dbd0a9299cab3451c1b3abdd35f6b5e76ca14a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:38:14.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:38:25.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: Ib070ffb0eb85ab0d5a299ca042a0646e8529fcac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/daff5843bff2b8f56f1fa9ba9643502f05355fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/daff5843bff2b8f56f1fa9ba9643502f05355fb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
