)]}'
{"id":"openstack%2Fblazar~907907","triplet_id":"openstack%2Fblazar~master~I1496b68436bb434ac2c34343564ce8e149ccf284","project":"openstack/blazar","branch":"master","topic":"compute_host_disabled_field","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-04-22 03:15:10.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2024-03-01 18:06:50.000000000","reason":"Change was marked ready for review"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-03-09 14:27:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"removed_from_attention_set":{"36611":{"account":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"last_update":"2024-04-23 15:19:46.000000000","reason":"\u003cGERRIT_ACCOUNT_36611\u003e replied on the change","reason_account":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"}}},"hashtags":[],"change_id":"I1496b68436bb434ac2c34343564ce8e149ccf284","subject":"Disabled field in computehost","status":"NEW","created":"2024-02-05 22:23:25.000000000","updated":"2024-04-23 15:19:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":42,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"a6d216082284d032c6726f600bd495bd142b1fbc","_number":907907,"virtual_id_number":907907,"owner":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-22 05:11:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 23:01:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 11:38:04.000000000","updated_by":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-28 03:40:32.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2024-03-08 14:15:21.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"4093aa3d9fbeedbfe2c7a67f56cc761f43fbc121","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-05 22:23:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8eda7c35a16ab886f31afdce595dfccd22989eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 23:01:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d982a4e16c44601a336e7ba40aecb65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bba74af6faf42dea4c92705f90640be : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d14fd69b0e4fb98152a86dc00a5055 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b7f5332b47e4fca9ecc6131b4b3c5ab : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bdd7dda0d4d4c32847fe70f60caa891 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebad61b17e3459ebb4b0f4ec9e91a1d : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa18c8c47bb84b99ae952f17ff931dee : SUCCESS in 3m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ada5d9ef6924475da80f142ef569a97f : SUCCESS in 35m 30s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f7655c4ebda442e4bfb0f30a979315fa : SUCCESS in 35m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5169314865949fdaaf26d637ec1b8b9 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ec4b45520cad1f1343bfe337e09ff637ebfcce3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-22 15:07:25.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ce6e44ebe729b7b7680ee004720c5b7536bd5882","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-27 16:00:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"437d4f2bf4a423545aaca2c96baa6fd28718496d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-27 16:00:43.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6f7c61067fab417ccab75a802c4225083fb66269","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-27 17:46:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce876908dd17a6b55bcb2452e6b319c55e5a2a50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-27 17:58:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7ee398392820728682c5ffb1417793056bb14f93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-27 18:05:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e77ed43e49e3062cba448359146fd51a8299b766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:50:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c0b391698ae429a83b95fc0c9f18d26\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ac7bddb44810450389217d0fcc2e3c5d : FAILURE in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b416020f1e34640a68fc30c3afc2925 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e44f7542e24ffea0264a7ac3d3f0af : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1fda303e68f479eaa42be716571bb94 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d3a0f5d248c4da59fc45ce011ebab79 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baad30e8b3cb42a2adcd70b9d4c46e50 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c74ea652ef85490f8940f5ae0121bb00 : SUCCESS in 3m 26s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b9646bb6651440f6a2e50b7a4af2f872 : FAILURE in 37m 45s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/27d48f85f5e64dacb0e4c993f7381e40 : SUCCESS in 39m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a859591c8b314243ba70b0ee531f7906 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1a9a5ffb4ce34b2398e30cc0e59e09ae162cb791","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-28 01:01:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"22542f65dd60556076e91db18475ab27c6c10545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:52:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dff5bf590b24cb18f9b3fe365aa80e6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2258584a324b4394a7b8556ae834a24e : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e90aaadc714a519d1e3506a6808e04 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e49a6f2d0b4de7935b5dcc8fdaa40b : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/639af102ca5e4c968175d71329f87836 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2fecd41aeaa4a0a899e62f278d8a5b4 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bfae88c5f440b6be9a6d74c10764f2 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/750272df8a1b48dd9719e3d86cd9274e : SUCCESS in 4m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/35977900886a46e78947dc024f7a0a90 : SUCCESS in 36m 57s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/af830dc5b9fd4b429f7c4b63154ab27e : SUCCESS in 44m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe6e6dce54664730ace42d5924456c99 : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c99bc63974a1e254814e9ecc5ca234b7e4f697cd","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2024-02-28 03:40:32.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8d1db540cdc4c603a8ed8c0a039d35f20c4b58b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-02-28 16:53:32.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"53191af6e227448d9cb28062d78c59ac30a22e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:37:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dffe119bb4d4651b0edbba3c6702e93\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f5062fbab66542f797a0cc2854b7073d : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc3c04b839ee4f4db7af0c49d5eed38d : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ee66c3b2bb4df399242b8897cc8f59 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad68bdb1be3a4072a094b793267b3ba9 : FAILURE in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4579b7dc601a473da01b4134ac8979e2 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c68767ce6fd45ffad675b0d4dc42b4c : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af419404f6d6496c9412fce27a3c811f : SUCCESS in 2m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/3acea8cfc79f41839b75f1e90df76f32 : FAILURE in 34m 11s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/bfcfed9effc74c09b25f38592d9a9f82 : FAILURE in 35m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f141ff6e33d3480e91a41ae80102a010 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"538d92061dbe75692475e5856f402a75b9e336ba","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-03-01 18:06:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"6e9eb1a9055ded8cafa3faeb6d24d69e95e2e74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-06 20:50:24.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_36611\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"}],"_revision_number":7},{"id":"9e999ead09c68e9ebfd3769d7e3817c60555db55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 21:30:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b19cd40ed65b465ea7f4f99b72033572\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41af02288edb4f2cb46aab506a93256c : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c77088ec1b44defad023f5a41bf5c8b : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc89142a30f4384a864a280323725ea : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c071f6b8879e47e9a4942c28aff915e0 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8431b35be2ca4d418cb872260905020f : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535d32b222d0491b9f98fc336a45480d : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f707b5b3e93b4679aec74d0a5baee0d6 : SUCCESS in 5m 13s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0bbc9aa2bd144591b65d5f60dd851dc9 : SUCCESS in 38m 29s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/159a88d2023d45a3a2b72f393a41d39d : SUCCESS in 28m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f3802d1a387404c87568c4d7b1f73f6 : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bcd5870ace18d9578466b28159c1d4fa11ad3f6c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-08 07:18:54.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"30c1ae08bf5b98ef1daa5fececa2074f481a46dd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-08 07:31:05.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f89da9d37dba61e1c0a1a55f42e3ac2210d1dab0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-08 07:31:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c5803210ca676d2224f886c442305522dab7eeca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-08 07:31:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fca44295985f845d1dd4d3db70970f09177e0f69","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-08 14:15:21.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"abb9af31fcdd78e351a3b7770c51e7676d253000","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-09 14:27:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b4a7837614475d43edd0764f05cfc743da0aea54","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-09 14:31:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"59f98c7f5d7989fd1f2a4a976f43973fe7065f58","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-09 15:02:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9f19a3467df3b4651178d61ccbeae55ac565834a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-03-11 03:17:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"fbe717f60d7c7b212975bd276e8aab9037ae4593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-03-11 04:06:04.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"3f7194a1d3e0dbaccbd463948f6ba34152f32026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 04:54:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8241d1cac7614c6c864129b4fa4ee859\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/032088d4897041ee9d4250e4ca0fe04d : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/764ec9d114f14663941dc5f64a05bb25 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2062561b543b4fc1aab9c55dd9c1224d : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/655ef026bebe4442a65a0048ea0cf1bc : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/956f3bd0ba074805899c79a19a1373dc : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62017db85dde4c4e87926969a62d3b54 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/489b4b422b8943c7b1bf46a2506a8926 : SUCCESS in 3m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/bff87a8cd9e0432ca1e3e00e248cf04a : SUCCESS in 40m 01s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5c2aebf83c2d4fe88f65825783f7f7a7 : SUCCESS in 42m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/903a30a5edbf42fe9fcbab7032bdd53f : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b86e9c5ccbc669e23d0e5f6fae3ea8187ccab963","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-17 21:08:44.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b80f9e9ce7407df0e5b365c7b20c6419bc1598f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-17 21:31:32.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f4c40115c15b0fb0931dbf5a8dae0f91170c575c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-04-22 03:15:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"65f9e2120cf53c0199dbbf900a53be33976ac0f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 05:11:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6461db1374f447dd945b2c264709e99c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/381f3346d1e34fb098ed7c20ef47201e : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b228de341d7047809ede7be890863c99 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84395477e9884487a496f66d1ba8466d : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fa22a9433f84e8b87030a3bc73c8f04 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91ae9d28b2cd43b993e67bbf7f16dbbe : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/febe02af819a488ca5963e5a9cbb4e23 : FAILURE in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b62f57eb10c43629a254516d2a2aaad : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f08ce949a6b4ec6822c1a68c785cbaf : SUCCESS in 3m 43s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a67a01fa7e7943fa9cb6d6083c3f12f3 : SUCCESS in 43m 33s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1a43d948d13d444caf1774d83624ac73 : SUCCESS in 42m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7526eced33f5484bbc027e90c7718dc8 : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a6d216082284d032c6726f600bd495bd142b1fbc","author":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"date":"2024-04-23 15:19:46.000000000","message":"Patch Set 10:\n\n(16 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b5ac33f247454a91b1d798a33bc96ac3867bd79f","revisions":{"2fa7a9fa28a16761c07fead05717dff45c1db591":{"kind":"REWORK","_number":1,"created":"2024-02-05 22:23:25.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/1"}}},"commit":{"parents":[{"commit":"667a8bd782579a81fab0a5048a6d44fc2bfc7059","subject":"Update bindep package list for Debian Bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/667a8bd782579a81fab0a5048a6d44fc2bfc7059"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 22:14:13.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can check in `_matching_hosts` is a user is admin or from admin project\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2fa7a9fa28a16761c07fead05717dff45c1db591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2fa7a9fa28a16761c07fead05717dff45c1db591"}]},"branch":"refs/heads/master"},"b8ae28f14105b209ff0c7799026d3443d3099338":{"kind":"REWORK","_number":2,"created":"2024-02-27 17:46:49.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/2"}}},"commit":{"parents":[{"commit":"667a8bd782579a81fab0a5048a6d44fc2bfc7059","subject":"Update bindep package list for Debian Bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/667a8bd782579a81fab0a5048a6d44fc2bfc7059"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-27 17:35:06.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b8ae28f14105b209ff0c7799026d3443d3099338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b8ae28f14105b209ff0c7799026d3443d3099338"}]},"branch":"refs/heads/master"},"1e35f4b31b414e033b08de76337ab0dec2925a06":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-27 17:58:16.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/3"}}},"commit":{"parents":[{"commit":"9aa976ae13384da92f28d3812a50d39e0a8a3798","subject":"Fix InstanceReservations affinity nullable flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9aa976ae13384da92f28d3812a50d39e0a8a3798"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-27 17:48:49.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1e35f4b31b414e033b08de76337ab0dec2925a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1e35f4b31b414e033b08de76337ab0dec2925a06"}]},"branch":"refs/heads/master"},"e7169b6114fe7f6a7253630844ec0c6e304234c1":{"kind":"REWORK","_number":4,"created":"2024-02-27 18:05:33.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/4"}}},"commit":{"parents":[{"commit":"9aa976ae13384da92f28d3812a50d39e0a8a3798","subject":"Fix InstanceReservations affinity nullable flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9aa976ae13384da92f28d3812a50d39e0a8a3798"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-27 18:05:24.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e7169b6114fe7f6a7253630844ec0c6e304234c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e7169b6114fe7f6a7253630844ec0c6e304234c1"}]},"branch":"refs/heads/master"},"cc25f3c6a7ce077d9f2ff2d13847d223f06f8d40":{"kind":"REWORK","_number":5,"created":"2024-02-28 01:01:10.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/5"}}},"commit":{"parents":[{"commit":"9aa976ae13384da92f28d3812a50d39e0a8a3798","subject":"Fix InstanceReservations affinity nullable flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9aa976ae13384da92f28d3812a50d39e0a8a3798"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-28 01:01:05.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cc25f3c6a7ce077d9f2ff2d13847d223f06f8d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cc25f3c6a7ce077d9f2ff2d13847d223f06f8d40"}]},"branch":"refs/heads/master"},"bea99e896f113d49e7c671f88a99c363683cbc6c":{"kind":"REWORK","_number":6,"created":"2024-02-28 16:53:32.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/6"}}},"commit":{"parents":[{"commit":"9aa976ae13384da92f28d3812a50d39e0a8a3798","subject":"Fix InstanceReservations affinity nullable flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9aa976ae13384da92f28d3812a50d39e0a8a3798"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-28 16:53:25.000000000","tz":-360},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bea99e896f113d49e7c671f88a99c363683cbc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bea99e896f113d49e7c671f88a99c363683cbc6c"}]},"branch":"refs/heads/master"},"e3f279c90a22ccb607fdd889bb8e843dfb313522":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-03-06 20:50:24.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/907907/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/7"}}},"commit":{"parents":[{"commit":"59082738b3a6118a32c42f47a973c1fccfec0e72","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/59082738b3a6118a32c42f47a973c1fccfec0e72"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy","email":"2anishreddy@gmail.com","date":"2024-03-06 20:50:24.000000000","tz":0},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e3f279c90a22ccb607fdd889bb8e843dfb313522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e3f279c90a22ccb607fdd889bb8e843dfb313522"}]},"branch":"refs/heads/master","description":"Rebase"},"4109b80aaf61727b3651bf7e49a4f0954649fb09":{"kind":"REWORK","_number":8,"created":"2024-03-11 03:17:33.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/8"}}},"commit":{"parents":[{"commit":"59082738b3a6118a32c42f47a973c1fccfec0e72","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/59082738b3a6118a32c42f47a973c1fccfec0e72"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-03-11 03:17:26.000000000","tz":-300},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4109b80aaf61727b3651bf7e49a4f0954649fb09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4109b80aaf61727b3651bf7e49a4f0954649fb09"}]},"branch":"refs/heads/master"},"678fe5eec0bb2cf334d2b5e49f7d3844aa5263b0":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-03-11 04:06:04.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/9"}}},"commit":{"parents":[{"commit":"59082738b3a6118a32c42f47a973c1fccfec0e72","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/59082738b3a6118a32c42f47a973c1fccfec0e72"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-03-11 04:05:54.000000000","tz":-300},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dFalse`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/678fe5eec0bb2cf334d2b5e49f7d3844aa5263b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/678fe5eec0bb2cf334d2b5e49f7d3844aa5263b0"}]},"branch":"refs/heads/master"},"b5ac33f247454a91b1d798a33bc96ac3867bd79f":{"kind":"REWORK","_number":10,"created":"2024-04-22 03:15:10.000000000","uploader":{"_account_id":36611,"name":"Anish Reddy","display_name":"Anish Ravula","email":"2anishreddy@gmail.com","username":"aravula"},"ref":"refs/changes/07/907907/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/907907/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/907907/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/907907/10"}}},"commit":{"parents":[{"commit":"59082738b3a6118a32c42f47a973c1fccfec0e72","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/59082738b3a6118a32c42f47a973c1fccfec0e72"}]}],"author":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-02-05 21:29:11.000000000","tz":-360},"committer":{"name":"Anish Reddy Ravula","email":"2anishreddy@gmail.com","date":"2024-04-22 03:14:41.000000000","tz":-300},"subject":"Disabled field in computehost","message":"Disabled field in computehost\n\n- The idea is to enable operator to reserve a non reservable node and\n  perform tests\n- Only operator/admin should be able to reserve such nodes\n- Polling monitor should not automatically make the node reservable\n  if the \u0027disabled\u0027 flag is True\n\nSet disabled field True for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dTrue`\n\nSet disabled field False for a host\n`openstack reservation host set \u003chost_id/name\u003e --extra disabled\u003dFalse`\n\nWIP: Setting disabled field works as desired (only admin can set/unset)\nCreating lease on a disabled\u003dTrue node doesn\u0027t work as desired\n  (any project and user can reserve such node)\nExpected behaviour: Only admins/admin project should be able\n  to create lease on such nodes\n\nLooks like using context policy.enforce does not work if an explicit\n  policy is not defined.\nDo we need to have an endpoint to define a policy? Not sure if we can\n  check in `_matching_hosts` is a user is admin or from admi$\n\nChange-Id: I1496b68436bb434ac2c34343564ce8e149ccf284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b5ac33f247454a91b1d798a33bc96ac3867bd79f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b5ac33f247454a91b1d798a33bc96ac3867bd79f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
