)]}'
{"id":"openstack%2Fblazar~917639","triplet_id":"openstack%2Fblazar~master~I7507e87908e16b313fd624810d6f43cb73dddafe","project":"openstack/blazar","branch":"master","topic":"flavor_plugin","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-08-28 20:35:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2024-08-28 21:10:42.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-28 12:43:31.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I7507e87908e16b313fd624810d6f43cb73dddafe","subject":"Implement allocation_candidates for flavor-based reservation","status":"MERGED","created":"2024-04-30 10:58:05.000000000","updated":"2024-08-28 21:11:49.000000000","submitted":"2024-08-28 21:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917639-flavor_plugin","meta_rev_id":"9628b6eaacec35365f62cb38060847c7fdbc686b","_number":917639,"virtual_id_number":917639,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 21:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 10:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2024-08-28 12:43:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 20:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 11:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 14:58:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-06-28 15:54:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2024-07-05 13:21:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-08-26 20:37:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"837a12ec5b82bdb6576999899866d4afb872e2a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 10:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d67af436647ee511a9a209b72207800708783cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:36:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1a239af80d0471dac30cc82f5e99511\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/885e4d20695141b7992dbc4c785ce157 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ef92d50793a47a68835c77e48b325b4 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b92780c7874861adb3fd2c145a8785 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91db54725840435cba4a5365f0c572af : FAILURE in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/048ad242ba57410898a61f73c4a52205 : FAILURE in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61f306a0b4f44d448cf6b4fb5438104e : FAILURE in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3907478b7cbc4b708d16b5a74a012eb2 : SUCCESS in 7m 13s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/98499c0b43a146e8bce45e8ffc93e113 : FAILURE in 32m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/be6871d15ce94556b0294f833a28cfb9 : FAILURE in 31m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f1c52a1d22114702aa36c5de386b6178 : FAILURE in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73b26fbe3cba73dbc598b13c92dba064a89b229d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 11:40:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e76841b46affa1c260d516e875b0cf73dfa890d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 11:40:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d8c4f121f8e230a1f23e92ff9fb76da30f77f182","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 11:40:42.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1a729ff8b8f4bfe616dc78fae9763329f60712ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 12:35:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8201f6dfa914bc4af1f0f3f39a9760c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3fd7b912229466181f6c1548b3361c9 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a81528a3a8d4b9eabd87eedcebfba86 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150e46df495e4d11863588231fb0cebb : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc8f63cce4694183af55a89d06e8f2e5 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04d78f7cf82940248d50f17cb1b5207a : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c0f608c402f4ea3bedf686a4cabea39 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70977b487d8493dab80c04bf60f0691 : SUCCESS in 6m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/942f5f3cd5ea47069a520481f115fc60 : SUCCESS in 44m 15s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/d41744ad969c43b69da155b4cea6cca3 : SUCCESS in 46m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d16567a85a94082af01af16c7f4222a : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"729aa1f7bf710495ae98f9d85a2cc36e6fd503e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 17:25:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f6f1e3be5eaddf3b510ea67c62933c9cf1066f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-30 17:27:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d72abc1a9e7db205120a3feb016e722446ec3230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:17:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c957a8acfb604d8fbf17e9e005a35a71\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4632df0621314871a9678f7497c667a7 : SUCCESS in 1m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fe9fa89bf374ff3a13defbd548cfcbe : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74bfeddd8d14a0780568b5d9c09312d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8fdd9c92164e34984031935a039697 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13742690336e4e03b057a46d03c04e7b : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6b17eac1e034031b2e87a7eed19e87b : FAILURE in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07309cc11f8c40aeb20bbf7129ffe6f4 : SUCCESS in 6m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1330d680aba548869c6ec3ed10f8347b : SUCCESS in 36m 21s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/df890b5c44bb417f90827f86a511dbeb : SUCCESS in 44m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7fab7283d5ba4d859333144cd8fab9bf : FAILURE in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d59ae146a4928e6c54a7c098739962839ccf0b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-01 09:30:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3c8291dd88eb2e2031259580dce6b5dcdebbe44a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 10:26:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/003fbf2d6c664285acd79de41801af3b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/429fb85eb41c41468f241ce783f80a04 : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebf211527c834606b0cece184eb1051f : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb6b5f558b54080b675b7fa1e05a072 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9224f373fff449669941795bd704be42 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54d6a2de2af5480fb0228cb3c60b517f : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b94f69f0274a86841ab736e360d6f8 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbccdcb71e83422e9dbac8dde68495dc : SUCCESS in 7m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/bc5655a21499446b8c68bf1f189fc71a : SUCCESS in 55m 55s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ef2a29a9dda54d5394b5857f697eace4 : SUCCESS in 41m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89a3c2911ffe4d68835ad2fc9f09d02b : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c94ee8552d1e7ae6429b182f138dc96f3593181c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-02 10:15:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e510ff509f4ac3d4a4829f6da2e162956699381c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-02 10:49:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3129b32b0046646421a9fbd96a90a2baf37456ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-02 10:50:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93bf4d167b276d2151303567965e759d71143dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:42:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0b3ef1036464458b9a77e12ecc2b702\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd11458f1b4d4955ad9a47ceb59f4162 : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3780d8a558214dff8f8607bf307af5ed : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cb971de8c145828a6477820b1a5366 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/216247a6cf174ccdb67124e4f20bafaa : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc7a28385e73411f8e73a740d4d2ae1b : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1bc0cf90d7b40bd9df308ce44dc339a : FAILURE in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56308d045d09451c93e8b6ca5b5c1b0d : SUCCESS in 7m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/28b5420e317145dab919fa3da25fdf9a : SUCCESS in 43m 26s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c9be18a718b746bbab950fdef3da9182 : SUCCESS in 30m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4955084e56f41dbb0ce624cf91f0fcf : FAILURE in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3d46048a011ef425032259ede0beb81b4b6813f5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-02 14:58:51.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fc3feda2be2f373a41d524b340f083fd089e3b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-03 18:05:02.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2cc42289c09843f0641da6c88f6b47d11a0ee2b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-05-03 18:10:13.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"8df0f8b8b0eda2da75455f19af39cbbf5e4b6441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:06:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7bb5b5069674eaf940e0385025c5fc7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6aaca72e4c85429eaec05259a24497a3 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44df145f4023499886acbcb4ca983e68 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a4be8027b84c9c9b9baa5e72036f1e : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fe432f7a9804889a6b872a57b8c560d : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40b1067271224e76b1ce087627353605 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f4b2eb1daa040629be05db463b68897 : FAILURE in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ad9478605a47aab815814e418e25a3 : SUCCESS in 6m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/aa725e598f6a46f2bbb9673b20398491 : SUCCESS in 33m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b2668b5cf94db2b631cca4532c58df : SUCCESS in 49m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4da1496f93194614bd496bc48920f0cc : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1789dd4b0d5a83e3a3a2847ba358bb40d50f6f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-06-07 08:22:57.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"cc328cc4e0cd452929ad5d31d96c1d2f5294357f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-06-07 09:01:08.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"b6dd6bc9cca5b5ebb9ef8c91d3fd655177262909","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-06-28 15:54:07.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2212ade3f058e6aa1d8f615a17e97e8078679455","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-05 13:21:51.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a2783e50870c590ae115e75c806b549d5209001a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-05 13:26:40.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2d40b251f89c7367781ffe2815acc6f328470b03","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-02 08:01:26.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"dd2f6e3e7575ef711ee55e87a3b169fff07c8ee3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-02 08:22:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c52048b92f8c3b76c9e6af836de1e93ba259f72b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-02 08:36:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8b27215d0d5fbfe7475cc9374c5842c0c9354144","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-05 16:00:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7d293e3c684441bb609b92c89d09ce333de30cbe","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-05 16:02:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3ffe5a2716cb0ad8f625b456e1114b9c17d59b09","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-23 12:36:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"65e21e6e95086effc74bc887c8712f7c62d2ea9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 16:13:18.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"be8cdddb535ac2d99d4f350cbc7e19075418d34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:55:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62c549daeb094bbdae5fcfb97dd816de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf40b8a8d6ee467a858c54c838dcbc35 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3beaa1a003e84bf38bf33d67a2e2109c : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d110c34f01574ca3be21f1ed4bbf18d1 : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76dcaa40963b47ccb23321fd6c29a4d0 : FAILURE in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/051e1191df8642c1b6925c5f7df928b7 : FAILURE in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59becd861f094c5195a091b032c437b1 : FAILURE in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe721b5ed2d414590fcd322f37442ec : SUCCESS in 6m 03s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/63499f73e6974bf9b528d09af7324886 : SUCCESS in 40m 01s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0f1195d434c94a20ba8cd0f2a67272da : SUCCESS in 37m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae574f35e65d4a2cb9a0170187aebdd6 : FAILURE in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6c09f86b929b9d5c08591afbb38440ef37df7ac1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 11:00:44.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"472a80b087e6b9064dca842f205e2101e6fbf0b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 11:01:12.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"231983bdf2444598259e25f6b40803c4a6601e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 11:01:26.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"743affbe77fea52fb8f834f2151536624235ef78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 11:13:19.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"64eb3c1a581d791b54ceca59a7d5871d925b3ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 11:51:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8a6735ac7d4c7a9f3801827c2969d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e4d9b597390427fb1df4eddb69e59a4 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b39c591a3ebc4ea69d6c4fc4e0e17f2f : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb4fd71399f48108c47ef0eb7906357 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9692456b8d9f4c8599447f0a90dbb168 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cbd4a9ee5d04f929287bfec7c3034a3 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11f0f5194d684373bccadd9dde08a33f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69785242b8724cdfb555744c6cd7db10 : SUCCESS in 8m 11s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0aa37d63887749468faa88303e5aabb2 : SUCCESS in 35m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/abb054db1c5e4e2ca155935934d276c8 : SUCCESS in 36m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/238da1ba3e624eca8ec2db258726ba76 : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"edc22f846a52fd0bf86751dbc367d362dcf5c624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 14:16:18.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"0794e1443e608f20d19e52d20987119958c4e8e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 15:03:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89284f9df20430cbf987d8c4762c6d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a38d82b2ae3472fb67d07f8300461f3 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaa04a36bcbc49c395e8d42407b1f25b : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa00a06fd1504ce0a02efed37e8b3775 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50a1e1ff2aed4b55bbd1990606ed1c54 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55450b3c93054ad19f8bbbe426fbfbad : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1eb16b4e57e240478ae23f31018e7b2d : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435fedb0e5a8492d98819efe4a2c4759 : SUCCESS in 6m 43s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/02d2e6bbc2634f32bb5795ff7042775b : SUCCESS in 45m 53s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/803afa0601cc4642a5a1d3e5c71016db : SUCCESS in 33m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9118c0ce86394b8abf8dd8d2e6cdebce : FAILURE in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7648397afcbf08dcfee81de8e19ba725b05a3898","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-26 20:37:52.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"c3a357e73798b9fdf62f4369073f76d8d5c35ecf","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-27 08:56:23.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"2b52a57aa257166f6f4b7c50613c4564b910a359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 09:41:22.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"6c7733a6fd3f3d032d0df929b48f154014287a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 10:36:35.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da5befcdf9964ea19116c00cb4b203f6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a83aab736e544af912cccce0fb6f8d0 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d66c2a05bc0d44cc9562435236f9ffed : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1b7073baa945ce80b2ac03998afd81 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf08d2f090f47d484d5b0a5f5c13586 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64003ba2911a4797a5eba5f1505300e1 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa498bca1dae425db6c1b4a724615ca7 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27010ee44ad402da47a831325014efe : SUCCESS in 6m 45s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ab05723507f14e43bacf26f966a141d7 : SUCCESS in 40m 32s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f31936e9d28646d68c665df2065986b4 : SUCCESS in 45m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1852e42d9c4b4a40ae743731a1efa930 : FAILURE in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"682c3fc59d37c3ad30c64aad6a57a125e37d531b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 10:39:19.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"9efeeca839015b77e68d09bba0fff4d2ad4daa48","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 12:43:31.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"b62e4277ee2e2c87b3c7164bab20e87c0a8f7c40","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 20:35:59.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"3bf84148a95d424252aa1471f845940877fee564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:37:02.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"9b48863c139074127bd9d6b60e665609163c57a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:10:41.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5400a106a347b598a70fd4a5e8c23a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b318806e400c4a52a7f6b635a97225ae : SUCCESS in 1m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4770e7540b5148eb97ee27b1c1bf796a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/723271344c1f423083babab7be9a21b9 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5423b93b81c44c38f0aa8ac86ff97f7 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056c3e7054c645fd9ab67dc1c8362bfb : SUCCESS in 6m 11s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/17b912c74ce54333b5451e800d4f3aec : SUCCESS in 32m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/d5db9141d62f4669bebde29dab61c3e1 : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":16},{"id":"5d6d271776ffe2f00fe7fd6e5fd205829185b0cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:10:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"9628b6eaacec35365f62cb38060847c7fdbc686b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:11:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1edc07e7a4d4b84a00ea732d3ed5f24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0452a7316ace47e8ba0221ece6e73389 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"d5dadf29877b5b2ad875e48ea71a11774b63150c","revisions":{"31a23fd7932b94c0521a708fcc93e39c3a86529a":{"kind":"REWORK","_number":1,"created":"2024-04-30 10:58:05.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/1"}}},"commit":{"parents":[{"commit":"c9c6ae5b4627a52e2faa5cc7d59952f0e2fdc5be","subject":"WIP: Write inventory and traits into DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c9c6ae5b4627a52e2faa5cc7d59952f0e2fdc5be"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-30 10:57:38.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nThis is currently based on the dummy plugin,\nand is starting to pull in ideas from the\ninstance plugin, as required.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/31a23fd7932b94c0521a708fcc93e39c3a86529a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/31a23fd7932b94c0521a708fcc93e39c3a86529a"}]},"branch":"refs/heads/master"},"4f4c8956a207d957871a58442cadac101a9784bf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-30 11:40:37.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/2"}}},"commit":{"parents":[{"commit":"699427bfa8f1b00dc620481b6d8c3457af8da54c","subject":"Write inventory and traits into DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/699427bfa8f1b00dc620481b6d8c3457af8da54c"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-30 11:40:37.000000000","tz":0},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nThis is currently based on the dummy plugin,\nand is starting to pull in ideas from the\ninstance plugin, as required.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4f4c8956a207d957871a58442cadac101a9784bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4f4c8956a207d957871a58442cadac101a9784bf"}]},"branch":"refs/heads/master","description":"Rebase"},"500657afc65724813f2a8207ccd92e027cd6d8ce":{"kind":"REWORK","_number":3,"created":"2024-04-30 17:25:43.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/3"}}},"commit":{"parents":[{"commit":"699427bfa8f1b00dc620481b6d8c3457af8da54c","subject":"Write inventory and traits into DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/699427bfa8f1b00dc620481b6d8c3457af8da54c"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-30 17:25:33.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nThis is currently based on the dummy plugin,\nand is starting to pull in ideas from the\ninstance plugin, as required.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/500657afc65724813f2a8207ccd92e027cd6d8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/500657afc65724813f2a8207ccd92e027cd6d8ce"}]},"branch":"refs/heads/master"},"6e308ce6f1e5e765b234c9dc663b3894af8e6876":{"kind":"REWORK","_number":4,"created":"2024-04-30 17:27:24.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/4"}}},"commit":{"parents":[{"commit":"699427bfa8f1b00dc620481b6d8c3457af8da54c","subject":"Write inventory and traits into DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/699427bfa8f1b00dc620481b6d8c3457af8da54c"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-30 17:27:14.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nThis is currently based on the dummy plugin,\nand is starting to pull in ideas from the\ninstance plugin, as required.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6e308ce6f1e5e765b234c9dc663b3894af8e6876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6e308ce6f1e5e765b234c9dc663b3894af8e6876"}]},"branch":"refs/heads/master"},"c8128bfb69cefe38c67e26d47bdd69976a0857e6":{"kind":"REWORK","_number":5,"created":"2024-05-01 09:30:16.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/5"}}},"commit":{"parents":[{"commit":"699427bfa8f1b00dc620481b6d8c3457af8da54c","subject":"Write inventory and traits into DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/699427bfa8f1b00dc620481b6d8c3457af8da54c"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-01 09:30:07.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c8128bfb69cefe38c67e26d47bdd69976a0857e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c8128bfb69cefe38c67e26d47bdd69976a0857e6"}]},"branch":"refs/heads/master"},"1fc6a7070d8bfb825d6ccd92611ee61258f0a99a":{"kind":"REWORK","_number":6,"created":"2024-05-02 10:15:57.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/6"}}},"commit":{"parents":[{"commit":"17fc483ae92148d961a0e4fe21ea45a55249ec2a","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/17fc483ae92148d961a0e4fe21ea45a55249ec2a"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-02 10:15:44.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1fc6a7070d8bfb825d6ccd92611ee61258f0a99a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1fc6a7070d8bfb825d6ccd92611ee61258f0a99a"}]},"branch":"refs/heads/master"},"328472841b1be262eb169af44318f9338b99a680":{"kind":"REWORK","_number":7,"created":"2024-05-02 10:49:25.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/7"}}},"commit":{"parents":[{"commit":"17fc483ae92148d961a0e4fe21ea45a55249ec2a","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/17fc483ae92148d961a0e4fe21ea45a55249ec2a"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-02 10:49:12.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/328472841b1be262eb169af44318f9338b99a680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/328472841b1be262eb169af44318f9338b99a680"}]},"branch":"refs/heads/master"},"93005f8c0560cc0ae78a148609005c7f6d2b16c5":{"kind":"REWORK","_number":8,"created":"2024-05-02 10:50:54.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/8"}}},"commit":{"parents":[{"commit":"17fc483ae92148d961a0e4fe21ea45a55249ec2a","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/17fc483ae92148d961a0e4fe21ea45a55249ec2a"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-02 10:50:42.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/93005f8c0560cc0ae78a148609005c7f6d2b16c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/93005f8c0560cc0ae78a148609005c7f6d2b16c5"}]},"branch":"refs/heads/master"},"a65698b79c512ed3c100ef7a56c707854452aefb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-05-03 18:05:02.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/9"}}},"commit":{"parents":[{"commit":"6a62c6e7cf99efdaa5aef21bd08227717ed279ac","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/6a62c6e7cf99efdaa5aef21bd08227717ed279ac"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-03 18:04:47.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a65698b79c512ed3c100ef7a56c707854452aefb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a65698b79c512ed3c100ef7a56c707854452aefb"}]},"branch":"refs/heads/master"},"413b323fbac59164cbf35248fe189bf0c0f60b60":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-05-03 18:10:13.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/10"}}},"commit":{"parents":[{"commit":"fa9dc4794b2fe5252b5a980b23651bb432285bb2","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fa9dc4794b2fe5252b5a980b23651bb432285bb2"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-05-03 18:09:16.000000000","tz":60},"subject":"WIP: Add a flavor plugin","message":"WIP: Add a flavor plugin\n\nStarting to pull in ideas from the\ninstance plugin, as required, where\npossible reusing existing methods\nfrom host plugin and instance plugin.\n\nSo far just lookint at allocation_candidates\nworking.\n\nTODO: need a patch before this one to extend\nthe instance reservation records to include\nthe details of the resource request.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/413b323fbac59164cbf35248fe189bf0c0f60b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/413b323fbac59164cbf35248fe189bf0c0f60b60"}]},"branch":"refs/heads/master"},"bd2774e719946724cbbb36be24e2320f9ab1cd74":{"kind":"REWORK","_number":11,"created":"2024-06-07 08:22:57.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/39/917639/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/11"}}},"commit":{"parents":[{"commit":"fa9dc4794b2fe5252b5a980b23651bb432285bb2","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fa9dc4794b2fe5252b5a980b23651bb432285bb2"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-06-07 08:17:29.000000000","tz":60},"subject":"Implement allocation_candidates for flavor plugin","message":"Implement allocation_candidates for flavor plugin\n\nThis adds the flavor plugin, where users request\na reservation similar to:\n\n  openstack reservation lease create test --reservation \\\n    resource_type\u003dflavor:instance,amount\u003d2,\\\n    flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the instance reservation\nplugin, and where possible leans on some of the existing\nlogic, currently only filter_hosts_by_reservation.\n\nThe plugin is not currently useable.\nThere are follow on patches that implement\nreserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd2774e719946724cbbb36be24e2320f9ab1cd74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd2774e719946724cbbb36be24e2320f9ab1cd74"}]},"branch":"refs/heads/master"},"b0b45afa8d4abfa106e70a082f61458f427a0439":{"kind":"REWORK","_number":12,"created":"2024-08-23 16:13:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/917639/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/12"}}},"commit":{"parents":[{"commit":"8d0bd44c5a9b00f700eeb45cc29b16d83dffbc78","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/8d0bd44c5a9b00f700eeb45cc29b16d83dffbc78"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-23 16:11:22.000000000","tz":120},"subject":"Implement allocation_candidates for flavor-based reservation","message":"Implement allocation_candidates for flavor-based reservation\n\nThis adds the flavor.instance plugin, where users request a reservation\nsimilar to:\n\n    openstack reservation lease create test --reservation \\\n        resource_type\u003dflavor:instance,amount\u003d2,\\\n        flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the existing instance reservation plugin,\nand where possible leans on some of the existing logic, currently only\nfilter_hosts_by_reservation.\n\nThe plugin is not currently usable. There are follow-on patches that\nimplement reserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0b45afa8d4abfa106e70a082f61458f427a0439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0b45afa8d4abfa106e70a082f61458f427a0439"}]},"branch":"refs/heads/master"},"faab89a0daf4c5de661605dd510c24f350b038a3":{"kind":"REWORK","_number":13,"created":"2024-08-26 11:01:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/917639/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/13"}}},"commit":{"parents":[{"commit":"5e6bba9d405b5cac6174d802e85b6c5b16a16a01","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5e6bba9d405b5cac6174d802e85b6c5b16a16a01"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"subject":"Implement allocation_candidates for flavor-based reservation","message":"Implement allocation_candidates for flavor-based reservation\n\nThis adds the flavor.instance plugin, where users request a reservation\nsimilar to:\n\n    openstack reservation lease create test --reservation \\\n        resource_type\u003dflavor:instance,amount\u003d2,\\\n        flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the existing instance reservation plugin,\nand where possible leans on some of the existing logic, currently only\nfilter_hosts_by_reservation.\n\nThe plugin is not currently usable. There are follow-on patches that\nimplement reserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/faab89a0daf4c5de661605dd510c24f350b038a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/faab89a0daf4c5de661605dd510c24f350b038a3"}]},"branch":"refs/heads/master"},"b80bee359f73621693e9bdc447152670869b518e":{"kind":"REWORK","_number":14,"created":"2024-08-26 11:13:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/917639/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/14"}}},"commit":{"parents":[{"commit":"5e6bba9d405b5cac6174d802e85b6c5b16a16a01","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5e6bba9d405b5cac6174d802e85b6c5b16a16a01"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-26 11:13:11.000000000","tz":120},"subject":"Implement allocation_candidates for flavor-based reservation","message":"Implement allocation_candidates for flavor-based reservation\n\nThis adds the flavor.instance plugin, where users request a reservation\nsimilar to:\n\n    openstack reservation lease create test --reservation \\\n        resource_type\u003dflavor:instance,amount\u003d2,\\\n        flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the existing instance reservation plugin,\nand where possible leans on some of the existing logic, currently only\nfilter_hosts_by_reservation.\n\nThe plugin is not currently usable. There are follow-on patches that\nimplement reserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b80bee359f73621693e9bdc447152670869b518e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b80bee359f73621693e9bdc447152670869b518e"}]},"branch":"refs/heads/master"},"1418c4395e9cbd286e45506c7e071bd8c8512fed":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-08-26 14:16:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/917639/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/15"}}},"commit":{"parents":[{"commit":"4219ab3f6202c736bacb568ffc03e8e2c0bbef9b","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4219ab3f6202c736bacb568ffc03e8e2c0bbef9b"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-26 13:27:26.000000000","tz":120},"subject":"Implement allocation_candidates for flavor-based reservation","message":"Implement allocation_candidates for flavor-based reservation\n\nThis adds the flavor.instance plugin, where users request a reservation\nsimilar to:\n\n    openstack reservation lease create test --reservation \\\n        resource_type\u003dflavor:instance,amount\u003d2,\\\n        flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the existing instance reservation plugin,\nand where possible leans on some of the existing logic, currently only\nfilter_hosts_by_reservation.\n\nThe plugin is not currently usable. There are follow-on patches that\nimplement reserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1418c4395e9cbd286e45506c7e071bd8c8512fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/1418c4395e9cbd286e45506c7e071bd8c8512fed"}]},"branch":"refs/heads/master"},"d5dadf29877b5b2ad875e48ea71a11774b63150c":{"kind":"REWORK","_number":16,"created":"2024-08-28 09:41:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/917639/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/39/917639/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/39/917639/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/39/917639/16"}}},"commit":{"parents":[{"commit":"4b3289766b4eaebb8b247fa2a32648dd231846ca","subject":"Add host_resource_inventory_get_all_per_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4b3289766b4eaebb8b247fa2a32648dd231846ca"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-29 15:06:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 09:40:39.000000000","tz":120},"subject":"Implement allocation_candidates for flavor-based reservation","message":"Implement allocation_candidates for flavor-based reservation\n\nThis adds the flavor.instance plugin, where users request a reservation\nsimilar to:\n\n    openstack reservation lease create test --reservation \\\n        resource_type\u003dflavor:instance,amount\u003d2,\\\n        flavor_id\u003d\u0027e26a4241-b83d-4516-8e0e-8ce2665d1966\u0027\n\nThis uses lots of ideas from the existing instance reservation plugin,\nand where possible leans on some of the existing logic, currently only\nfilter_hosts_by_reservation.\n\nThe plugin is not currently usable. There are follow-on patches that\nimplement reserve_resource and then finally update_reservation.\n\nChange-Id: I7507e87908e16b313fd624810d6f43cb73dddafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5dadf29877b5b2ad875e48ea71a11774b63150c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5dadf29877b5b2ad875e48ea71a11774b63150c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
