)]}'
{"id":"openstack%2Fblazar~924111","triplet_id":"openstack%2Fblazar~master~I150e3f0caad1ab5196d54e75d0ff366ec66797a7","project":"openstack/blazar","branch":"master","attention_set":{"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2024-07-13 11:58:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I150e3f0caad1ab5196d54e75d0ff366ec66797a7","subject":"WIP: add dry_run option on lease create","status":"NEW","created":"2024-07-13 11:36:15.000000000","updated":"2024-09-30 18:35:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"478e9abf6b78dfd5ec32d82c74b141c61219c154","_number":924111,"virtual_id_number":924111,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-30 18:35:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-13 11:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adbfbebb8f93070d4bc3f1c5db8e4f5ba2009547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-07-13 11:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20eaf06aebb3054a3e3578a033bf398d3c2380bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 11:58:06.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4cea0da3e3847b1bdc9effc42438144\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/151c3f1ce521407088fef28e2bf0693b : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1302f45cc4b944918701b9d977e44b40 : FAILURE in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfafa1b540064d85a4f7a516e9d3502b : FAILURE in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97b9e16439054422b2bc84250849d64a : FAILURE in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/463c4d411c9340d082b6313edd2a62e1 : FAILURE in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12af0c55e5649d88596431baef15c08 : SUCCESS in 5m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/99c73d6554534ee7b143725017edec7d : FAILURE in 17m 15s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/69d5378fb6624ce181cc5995302f0323 : FAILURE in 21m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2c6ae16655647e1aaf42e3f2a582b0d : FAILURE in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae1220b130b93f7634c1f018c63b5b3d85236a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-09-30 17:55:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"478e9abf6b78dfd5ec32d82c74b141c61219c154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 18:35:41.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6766ec783cd340d1bbdaff20ac861314\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3ffa59ce974e2fb0596a55a8a912ec : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a89363a1a84f6ab296915ec91c648c : FAILURE in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67cbf7a1fe1462f943b5b5ad90a6079 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d52b9ee3e9641aea75bb592708c417e : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04314b840214e82959fa2595643414f : SUCCESS in 6m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/146ab69a11874fa790b472ccbe9bc1f8 : SUCCESS in 38m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b41208c15c474293b322097d1f05dde5 : SUCCESS in 36m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a236381dd56c49d8bf886d4377162354 : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"097ba6746061f4978305f05f6e92800c69014218","revisions":{"f0d8db5eb7b0311383c7dac247a345d6e4840501":{"kind":"REWORK","_number":1,"created":"2024-07-13 11:36:15.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/11/924111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/11/924111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/11/924111/1"}}},"commit":{"parents":[{"commit":"b7d4b086242b133009721a6790e0fddce83f6a6c","subject":"Fix up read session handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b7d4b086242b133009721a6790e0fddce83f6a6c"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-07-13 11:26:34.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-07-13 11:35:59.000000000","tz":60},"subject":"WIP: add dry_run option on lease create","message":"WIP: add dry_run option on lease create\n\nOption to just do the enforcement check and resource allocations check,\nbut not actually create the resources.\n\nChange-Id: I150e3f0caad1ab5196d54e75d0ff366ec66797a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0d8db5eb7b0311383c7dac247a345d6e4840501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f0d8db5eb7b0311383c7dac247a345d6e4840501"}]},"branch":"refs/heads/master"},"097ba6746061f4978305f05f6e92800c69014218":{"kind":"REWORK","_number":2,"created":"2024-09-30 17:55:59.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/11/924111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/11/924111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/11/924111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/11/924111/2"}}},"commit":{"parents":[{"commit":"db69073828f76dadc2a7d915484ee1f2a6f9bcbc","subject":"Add extra on_end calls on reservation failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/db69073828f76dadc2a7d915484ee1f2a6f9bcbc"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-09-30 12:46:04.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-09-30 12:46:04.000000000","tz":60},"subject":"WIP: add dry_run option on lease create","message":"WIP: add dry_run option on lease create\n\nOption to just do the enforcement check and resource allocations check,\nbut not actually create the resources.\n\nChange-Id: I150e3f0caad1ab5196d54e75d0ff366ec66797a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/097ba6746061f4978305f05f6e92800c69014218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/097ba6746061f4978305f05f6e92800c69014218"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
