)]}'
{"id":"openstack%2Fblazar~930806","triplet_id":"openstack%2Fblazar~master~I4d3382b142c1c9225cad29f9e33abf539ecf0f30","project":"openstack/blazar","branch":"master","topic":"replace_deprecated_datetime_utcnow","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2025-02-06 15:55:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d3382b142c1c9225cad29f9e33abf539ecf0f30","subject":"Replace deprecated datetime.utcnow()","status":"MERGED","created":"2024-09-28 12:09:52.000000000","updated":"2025-02-06 15:56:07.000000000","submitted":"2025-02-06 15:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930806-replace_deprecated_datetime_utcnow","meta_rev_id":"697045b82e88fa8ef43a1eab017b87a28430b271","_number":930806,"virtual_id_number":930806,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-06 15:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-06 15:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-02-06 15:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-06 15:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-28 12:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-06 15:10:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-02-06 15:13:56.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"}],"messages":[{"id":"132a7e25f715dd3c97b9f9e4892722e3c4afcb25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-09-28 12:09:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb3f9471a4704e32f510d0ed13cb493d118db5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 12:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e11cbdf0606149e0bb4764a4573a43d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f724081aee24e54a1bd4c02ea91a1b0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f3dfb81d274fb09bef8afb3e989f62 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a970e0fd6874d3ba3982c7776c279b5 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab5b2cde92304088b90085d03fffe898 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02db5817e57640debd56a601df4ac1b8 : SUCCESS in 5m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a385b29e3bf34ddd85f1c099aab07dac : SUCCESS in 24m 39s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/83a6ab5e81ce42dfbf17799b6ce0303c : SUCCESS in 24m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d34468d171fc44039c2cededf993189b : FAILURE in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee9a001aff05a9ac8d540bb2c69dd0c0b6241c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-09-28 15:36:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"59dbeec371567c0495240abe0e766539becf1c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 16:02:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d239c92fee94c569b459c1748d90abd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0069bfa5b8424a649ac76217894ecb20 : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acee84d679b34851a96a867af94a42aa : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b2fe1c42b946ae95e3a3ad4645fe64 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211b5c4798a446ceb16248df1b71b35a : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9dc209a0a3644d1385760ffd0c84674e : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd1311ee65454747aab7283e4f44a5e8 : SUCCESS in 5m 16s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b858e50c81114954afa854e860b5962e : SUCCESS in 24m 56s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4ab0f912bebf4a4cbaf3d5183449390f : SUCCESS in 25m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18f1437b95a94e57863b4d61db42a7a8 : FAILURE in 2m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3cad47d59e512a1a799a3797aec3da23b6a7d09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-12-07 14:27:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a103d20d368d9c1b5e6238755a6d2d301cc53e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 15:09:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b3a91c28bb41e3b91d88b3190466c0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0fc2fc5ab814c158230bfd6e6aa72de : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e4c5413681d477881517d3d95b05043 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53bf6e6112234769b804ab684d4882a6 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97426140d4eb4700afb299b54e3b46a3 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1800eda005814e41934e85b7c8858c29 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19f3f8a0308410ea6b4523bb6a391ed : SUCCESS in 5m 32s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7c4122146c2742129b51507514ac9b78 : SUCCESS in 28m 47s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/a636c9125e0b4473860fe82c6f386c61 : SUCCESS in 41m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/13a5f9759f3e447cb8bde97eb53e1512 : SUCCESS in 26m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c62fdcf813b44c5bb3bf1c9242c932dd : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c8e4f4998da0d4c3c3a7cabf13656ecb777815a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 15:10:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2cd384da2cec544c30b409afb357ff226c0e0586","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-02-06 15:13:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8374d4614ca93d594b82996ae503cf76d9b0cea3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 15:14:43.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0f0322fe539f05fa272f89aa7e9fecf28a386f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:15:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5bf302b9e476a7775735ebb1ce16d12a4d331e10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:55:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed33e28f2d574202ad98ff14bd516cb2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bbfe5b2690e498a935d603d8ac09474 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e257b250a1a943919c3d54b07fcb9be7 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fa9525c741e4b15af52123de7d5073a : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c027454f8d13461a9a04867d376d71d0 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f9667885744ff28f097dce4db00798 : SUCCESS in 7m 15s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/37b14f6123ca4afc89b1d8cd37a91caa : SUCCESS in 36m 11s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/6a05fb015bd34771945582625dae1d3c : SUCCESS in 35m 20s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a67e86fd5b0b4ed1a87115e92c6b15fc : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6754850d7c1403089455ab7888519b1fbe72c0b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:55:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"697045b82e88fa8ef43a1eab017b87a28430b271","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:56:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b32745c9b574404a4b15acaaa4612f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23b9b0aa7d3476b80d258774d57efad : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5df5277756a6d832bcdf0107ce9df859ed2a1b03","revisions":{"916c0d49775cbbd8d53ee1230a865645c69cf83f":{"kind":"REWORK","_number":1,"created":"2024-09-28 12:09:52.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/06/930806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/930806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/930806/1"}}},"commit":{"parents":[{"commit":"3483be483d4705f3d19f0f96c65eb4130cdeeda7","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3483be483d4705f3d19f0f96c65eb4130cdeeda7"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-28 11:45:53.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-28 12:05:50.000000000","tz":540},"subject":"Replace deprecated datetime.utcnow()","message":"Replace deprecated datetime.utcnow()\n\nThe datetime.utcnow() is deprecated in Python 3.12.\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow().\n\nChange-Id: I4d3382b142c1c9225cad29f9e33abf539ecf0f30\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/916c0d49775cbbd8d53ee1230a865645c69cf83f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/916c0d49775cbbd8d53ee1230a865645c69cf83f"}]},"branch":"refs/heads/master"},"52a2937f6b11a1588ec09a8d75898a64a2b4b74b":{"kind":"REWORK","_number":2,"created":"2024-09-28 15:36:06.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/06/930806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/930806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/930806/2"}}},"commit":{"parents":[{"commit":"3483be483d4705f3d19f0f96c65eb4130cdeeda7","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3483be483d4705f3d19f0f96c65eb4130cdeeda7"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-28 11:45:53.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-28 15:34:29.000000000","tz":540},"subject":"Replace deprecated datetime.utcnow()","message":"Replace deprecated datetime.utcnow()\n\nThe datetime.utcnow() is deprecated in Python 3.12.\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow().\n\nChange-Id: I4d3382b142c1c9225cad29f9e33abf539ecf0f30\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/52a2937f6b11a1588ec09a8d75898a64a2b4b74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/52a2937f6b11a1588ec09a8d75898a64a2b4b74b"}]},"branch":"refs/heads/master"},"5df5277756a6d832bcdf0107ce9df859ed2a1b03":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-07 14:27:25.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/06/930806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/06/930806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/06/930806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/06/930806/3"}}},"commit":{"parents":[{"commit":"896ef7f935573dbf297eb47e504487fad3180012","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/896ef7f935573dbf297eb47e504487fad3180012"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-28 11:45:53.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-12-07 14:27:25.000000000","tz":0},"subject":"Replace deprecated datetime.utcnow()","message":"Replace deprecated datetime.utcnow()\n\nThe datetime.utcnow() is deprecated in Python 3.12.\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow().\n\nChange-Id: I4d3382b142c1c9225cad29f9e33abf539ecf0f30\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5df5277756a6d832bcdf0107ce9df859ed2a1b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5df5277756a6d832bcdf0107ce9df859ed2a1b03"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
