)]}'
{"id":"openstack%2Fblazar~942014","triplet_id":"openstack%2Fblazar~master~Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9","project":"openstack/blazar","branch":"master","topic":"heads/openstack/master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-27 12:22:32.000000000","reason":"Change was submitted"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2025-02-27 12:22:32.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-02-27 09:56:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9","subject":"Fix host randomization","status":"MERGED","created":"2025-02-17 22:32:21.000000000","updated":"2025-02-27 12:23:33.000000000","submitted":"2025-02-27 12:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942014-heads/openstack/master","meta_rev_id":"87c07870dc823fb697b3315486372ca76d0b915b","_number":942014,"virtual_id_number":942014,"owner":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 12:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-27 09:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-02-27 09:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-02-27 09:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 23:25:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 11:04:05.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-21 14:05:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-02-27 09:48:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"d4f8b0190b422d7c0e9e4c062551412a67dfbb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-02-17 22:32:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acdab57ce52089ce25e42d7e584064d094526222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 23:25:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f162aaccf8484e99debad3dad900b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d04d4ec49edf45de98ead282b210d588 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e0b446b73e43bfac65a24711c5162a : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca20ba47d34344c7a4dd03b01c3b0ec2 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cf5cbf1383142f29914b3ae8e878911 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26c1b4952224eda9b2ab2f094721c34 : SUCCESS in 5m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ef07bbfd3f2e4608884e240d0b2cfda6 : SUCCESS in 26m 19s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/07c2cb9926884262a47406661148e384 : SUCCESS in 41m 28s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/693c4c8b28b54faea038388ee4f22f10 : SUCCESS in 46m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59f93c0506d44ce7ac406b522a1d41e8 : FAILURE in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcd3f231ff97a272bf95fca8651cc761ca8d6811","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-19 11:04:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c389493f69fb65f14a491f4ab8fc12000b934680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-02-20 17:05:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"67e4c24df67192394bdf39f4b0ee761ae9b3500f","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-02-20 17:06:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d54ff89ae8f7cdd6fcd6e68eb1e26e9c38c155f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:50:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0eb30ff24cb45da8ea3d3cd9ecfdd62\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f7f9a03b6c4a0b879c6f015f897f53 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6feed69ba145b0828d1375e1ea1fb2 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08a8f7e2ecb0436680fe1a963e5432f9 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7efc54543dfe409986e80034f03f1b73 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f47df3eb22846b3a2ab6a71858e04d0 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb333a1dc3d34fc082762dfa89e33565 : SUCCESS in 3m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5afc804de85e41c2ae97795b74e6ebcf : FAILURE in 35m 30s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/ace061d3d22845b6b8e89b7834a37a62 : SUCCESS in 40m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1f08ddf236514c81bce2c3cf561f952d : SUCCESS in 42m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f5f665da247431c9675a895234e4365 : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e6ad07756e974520b57b8805443410ae4d3545c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-21 14:05:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"985d8d864d144a926a2b99b33cc3481f2dbbc82a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-02-24 18:57:26.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bfaa951bc9aaec3807d154bbbc7736bdd3f91037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:55:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8724f800832241f189b595a9a89ac115\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d55251a18de44da491c53b5a730d2a88 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbba8207bb74f9e96f39bb2a49db3b5 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7433775b69d45f7852a90cd50702175 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a919a7012d49c7beef3df00574243b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad0d5e062fe4585b5663e22cfa3f7b4 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed6eb05cc1374524b4e36d8fc6fd5bff : SUCCESS in 4m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ebbd80dc90244c10a6c5d2b3a5c86dd9 : SUCCESS in 48m 29s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/73519db7eeda4866893d8914cba626ba : SUCCESS in 49m 45s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4d93dd9e405e4615ae278ee21ec7e8f6 : SUCCESS in 46m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87adda2638294d9783fff3a2722e4dd3 : FAILURE in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3e1cdd29fb34816ba24d15019ab1e8deba00b8e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-25 10:27:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31e25aaac4d313a1a2cb4aa63e8a4b5d6573a616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-27 09:41:59.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3836e2e7351f6c22d32844b9b5909211947b2765","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-27 09:43:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70ab5cc35b6f81e86905a25a4d660de553212ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-27 09:46:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76a47410b4b98b045a05c6d85ef3d0fb705fafbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-27 09:48:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2de5edb8a9bac270ae6eb5207dd65f4ac52871e3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-27 09:56:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"20dad1aa27af127a0aac05ef2a366a8eba050b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 11:26:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1687ecf8dda41878ba87d31fba2d5a3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686826715742433e90378a57aefeed95 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3de3ed32874eae9c14fcd267f001f0 : SUCCESS in 17m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52251073f30b46a4a5a2ee732ba6ebe5 : SUCCESS in 20m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44ec2775c002486693084fabd3d9f2d3 : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7deb5f147624391ba3cce2004d8bbb7 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8de1eb273f64906a16582f421948b22 : SUCCESS in 3m 28s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/95e93933709d433697b2169ecec84903 : SUCCESS in 51m 06s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/e6243ba0025b4e44a3808e10e62b3a7f : SUCCESS in 47m 37s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/d43d4cf6615346729438cb1902432a8b : SUCCESS in 1h 29m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b17a98cd77b94d089b849fe532e1e3e4 : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43fb0fab49e30d6e8c51c90a40015371af9926c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 11:26:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c161a970047b179b6315ca5e9d44032e2ba2aada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:22:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe2ec7a41d54efeaf1d8b2e68598b79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f35440b4bb34370b09dc0ac30b3028d : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe71f961cb63404da7275342c4e7fe39 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cf03b9bc07641c5ab266eb58fcccb70 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9b5b99b52924879a453064285c31a0f : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e983f9d7b3a4f16ac1fa3b69d709781 : SUCCESS in 3m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1d1a70e77c0a4966be77afd9803986b3 : SUCCESS in 29m 16s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/7fdc983143f04f36973f7dd4d11f9858 : SUCCESS in 47m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e229e1dafc084a7287ac7832dfc8b7a7 : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":5},{"id":"1da4077c0c8e8ef801bc965c578c049b1e6ff17e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:22:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"87c07870dc823fb697b3315486372ca76d0b915b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:23:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07e51f31183d4fd8999a13fbb5d66108\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553b0fac22a243c99d8bba50865ab9a6 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a442f46879b0461e8e309e9c277229a5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dbdf1b894ff898bf4909789704120056677f3796","revisions":{"35550f4285296223ff7e7c64e8fe574d8a46be0d":{"kind":"REWORK","_number":1,"created":"2025-02-17 22:32:21.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/14/942014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/14/942014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/14/942014/1"}}},"commit":{"parents":[{"commit":"762dacbd3b655f777bd4a68be40479e95bd0fe41","subject":"Merge \"Replace deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/762dacbd3b655f777bd4a68be40479e95bd0fe41"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:20:50.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:32:21.000000000","tz":0},"subject":"Fix host randomization","message":"Fix host randomization\n\nPreviously, this was implemented using the shuffle method of the Random\nclass without instantiating a Random object, which did not work.\nThis changes swaps to the random.shuffle method, which works as\nexpected.\n\nChange-Id: Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/35550f4285296223ff7e7c64e8fe574d8a46be0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/35550f4285296223ff7e7c64e8fe574d8a46be0d"}]},"branch":"refs/heads/master"},"3a079d8d0c14af23c076f37c60254a53c500aa9c":{"kind":"REWORK","_number":2,"created":"2025-02-20 17:05:51.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/14/942014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/14/942014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/14/942014/2"}}},"commit":{"parents":[{"commit":"762dacbd3b655f777bd4a68be40479e95bd0fe41","subject":"Merge \"Replace deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/762dacbd3b655f777bd4a68be40479e95bd0fe41"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:20:50.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-20 17:05:17.000000000","tz":0},"subject":"Fix host randomization","message":"Fix host randomization\n\nPreviously, this was implemented using the shuffle method of the Random\nclass without instantiating a Random object, which did not work.\nThis changes swaps to the random.shuffle method, which works as\nexpected.\n\nChange-Id: Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3a079d8d0c14af23c076f37c60254a53c500aa9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3a079d8d0c14af23c076f37c60254a53c500aa9c"}]},"branch":"refs/heads/master"},"9b10e5477a18ab6ddfddc0c1c9f3ed27da0da98a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-24 18:57:26.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/14/942014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/14/942014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/14/942014/3"}}},"commit":{"parents":[{"commit":"762dacbd3b655f777bd4a68be40479e95bd0fe41","subject":"Merge \"Replace deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/762dacbd3b655f777bd4a68be40479e95bd0fe41"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:20:50.000000000","tz":0},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-24 18:56:55.000000000","tz":0},"subject":"Fix host randomization","message":"Fix host randomization\n\nPreviously, this was implemented using the shuffle method of the Random\nclass without instantiating a Random object, which did not work.\nThis changes swaps to the random.shuffle method, which works as\nexpected.\n\nCloses-Bug: #2099927\nChange-Id: Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9b10e5477a18ab6ddfddc0c1c9f3ed27da0da98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9b10e5477a18ab6ddfddc0c1c9f3ed27da0da98a"}]},"branch":"refs/heads/master"},"f7d7e25e37e01055e776a24bd440547c119fa796":{"kind":"REWORK","_number":4,"created":"2025-02-27 09:41:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/942014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/14/942014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/14/942014/4"}}},"commit":{"parents":[{"commit":"762dacbd3b655f777bd4a68be40479e95bd0fe41","subject":"Merge \"Replace deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/762dacbd3b655f777bd4a68be40479e95bd0fe41"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:20:50.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-27 09:41:53.000000000","tz":0},"subject":"Fix host randomization","message":"Fix host randomization\n\nPreviously, this was implemented using the shuffle method of the Random\nclass without instantiating a Random object, which did not work.\nThis changes swaps to the random.shuffle method, which works as\nexpected.\n\nCloses-Bug: #2099927\nChange-Id: Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f7d7e25e37e01055e776a24bd440547c119fa796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f7d7e25e37e01055e776a24bd440547c119fa796"}]},"branch":"refs/heads/master"},"dbdf1b894ff898bf4909789704120056677f3796":{"kind":"REWORK","_number":5,"created":"2025-02-27 09:46:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/942014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/14/942014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/14/942014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/14/942014/5"}}},"commit":{"parents":[{"commit":"762dacbd3b655f777bd4a68be40479e95bd0fe41","subject":"Merge \"Replace deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/762dacbd3b655f777bd4a68be40479e95bd0fe41"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-02-17 22:20:50.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-27 09:45:58.000000000","tz":60},"subject":"Fix host randomization","message":"Fix host randomization\n\nPreviously, this was implemented using the shuffle method of the Random\nclass without instantiating a Random object, which did not work. This\nchanges swaps to the random.shuffle method, which works as expected.\n\nCloses-Bug: #2099927\nChange-Id: Ib288091af9cc035ccb0535fbc1748c17bb3cb1e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dbdf1b894ff898bf4909789704120056677f3796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dbdf1b894ff898bf4909789704120056677f3796"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
