)]}'
{"id":"openstack%2Fblazar~949486","triplet_id":"openstack%2Fblazar~master~I190daa89d134a4440dd914a9a935000aaa87d4f2","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-05-22 10:09:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I190daa89d134a4440dd914a9a935000aaa87d4f2","subject":"Fix lease date","status":"MERGED","created":"2025-05-12 10:57:11.000000000","updated":"2025-05-22 10:10:20.000000000","submitted":"2025-05-22 10:09:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949486","meta_rev_id":"1ec26097c91a4cbfb9250aee3effc8c3a476f89d","_number":949486,"virtual_id_number":949486,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 10:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-12 11:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-16 08:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-16 08:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 11:09:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-12 11:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 07:57:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2025-05-16 08:39:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"ecce87c85b570cc60a2f5dc87ef5cb0f752301af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-05-12 10:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b089dcbdce8942865f66a37219abf62e11583425","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 11:09:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2ddb58982eb51f7a8b639fc9058111d8c32e2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 11:31:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2c171a073944b4e9477565cbbc67ab1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c24440ed3ad344c3892e66147be8ed1f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d07aaa24f04094afb545953cecae81 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e80bec7e6c0f4faf85adedb648e8a770 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ef4d40de1fb4bf49ebfa071765f31f4 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc7d0b57cdae4d5987ac36ce19fa4e8d : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a52d9286204c2a8c2a263ece297178 : SUCCESS in 7m 32s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e8df0f761231429596fa845a7f29357d : FAILURE in 27m 37s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/7cc7aabb23b04925a6bad9e12c65849c : FAILURE in 27m 34s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ac4db8dd05c54fffb92dc08cb915fd81 : FAILURE in 27m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/56713a1b642d4f42af7047cfdb0b411d : FAILURE in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ef5264f4b9617fde80e82b687c26aa70c0c7334","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-16 07:57:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67cca1f30c285046e2ed035e2f64447d6e3a011f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 08:35:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff75295491434f01bb0773cf9e320082\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93e503022ed043208fbe9d542b7c4a80 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c165851fd6414a84aa559164da74bb : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b836f3a3a16d40e28bfb2c323526f9d1 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4ec6214c0bd4d59ad6f5809dcf06432 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5cdd90d71104b1abe7102f4db2a11c7 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52c15c7def941088545d700544da76e : SUCCESS in 8m 25s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b2998725e70e49c6b36ef83db5eb5ed1 : SUCCESS in 32m 38s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/eed26a9b348f49fe905122eef3762f71 : SUCCESS in 21m 51s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/14485eb1f2034e70872cf45e93ffe069 : SUCCESS in 37m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50e979f1489c4cdc8eae18d8eb21c634 : FAILURE in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b716dea79e8072bc512f06a436331fb857b31fa","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-16 08:39:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b35a85a27e26797e4fa71c457513d34e6834b169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 08:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"742b7577297ee661f04b6e02e46f0e1d67e87fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:22:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b69fad2119d44f2a7c7396333eaaf98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6866bbaa418e462b8edae67d2efcdde3 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6518e6df1ae5417182649af2838e4319 : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1f07c565a8144cab3ce874d514564e4 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c45a3e8528e4e4ba97bfaffaecb84f4 : SUCCESS in 9m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5796eecbc19d4e5a82af41cf7a2e9ace : SUCCESS in 35m 04s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/c302831e40134e50940f8cd0bbc63ea3 : FAILURE in 38m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ef83423bf3684f40b94699f34fc59556 : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4ec50b19af63f99d841648ac9ab22b79a5e5ae0a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 09:25:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c32282958365c272de4debd815e3bdadd5515965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:51:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/346de176c5684aa29130b81f13091006\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7e797d98cae43a09425c8e7b1f5285c : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6831f1944604b0fb0d35d1e3597039b : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b2662e38690419082cddc1da46de9e8 : SUCCESS in 6m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29f06c0a419d4e1ebeec2c5a2a948018 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41eecf63d2f04c0cb5c84f0ec1879624 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa09ea3c63e4a4392b239f251ab44a7 : SUCCESS in 9m 43s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4d6c1b72ce884504a5de8e4b6304d324 : SUCCESS in 24m 33s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/d0bc94c024244000a0eabcf0f0445033 : SUCCESS in 21m 29s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/cc0540234a7d4784b0767c265b1383e7 : FAILURE in 11m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/018cf44adf1b43f187bbbf13b36488e0 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e1651dc369f2f27790fce1fa24820266f544f52","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-16 13:32:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77cd8b53babbe618b81cf087f6c9fa4d1ca66699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:11:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1454ad239b5b4aa39ea65adc237e4c63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb00edc636144358e60ded6cc04ebd8 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f344be45a1f43af8ba9f8e49de1852f : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc3b535ef4204127b23ca5c2613524b4 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2e6ce7e4df341629d1d73a2d57f324a : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26387c020adb4ecf96bce09ecba0f433 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415f36a821954e879c134355a4ad0bc0 : SUCCESS in 8m 08s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/6cd777fdd9dc47e8b41787671777effe : SUCCESS in 37m 23s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/8eaf65bf5405402580e98eb6105e694c : SUCCESS in 33m 47s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b0dd52c3c95a4b97af528b2eb9bb0b3c : FAILURE in 30m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfc84676efa14ed38dcd09db64295bb5 : FAILURE in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bd6f7f47915641fc08545ce1c4f5ab95aa191f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 15:02:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f659ad360121b2c742474f8329051ff6ef8d1d61","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-22 08:35:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc77b5d23b893c2da43626902e2f148823b79489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:19:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0cf411e6df48d4812fd0c7d0015764\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a571b991c38e4818aefd65b5471b2e7f : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3057f55b5ad14ea4a561e03418b22e80 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32b5033e6b3645d08aeee6f405be1664 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdbecfdf333348febf773637d1ae389c : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30030a892b1f447f8c7b6b88f5efdd6f : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96eff2b24c8440588952c592fded953d : SUCCESS in 8m 34s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4fbe71643766488382ebaa30149b5813 : SUCCESS in 38m 49s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/e6ba2a803e644496bcacde0e2f67b705 : SUCCESS in 37m 07s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a9eb9e80b61844349a4f47e07566689d : SUCCESS in 42m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd4fce5a26b84ecfabd1a95994d76fda : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11477ea22f5946d57ade638185a236001041e92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:20:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfd9b73cf3423e2714b58b1eb2fe2f168f4bd162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 10:09:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540f8014e60842ebab0799a30c80b4d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1612e39eac46e68bcab6fdf3c5a9f0 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57c106b67dba4aeca71ce0ee958c2f82 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d870ee7082b14c23a699274b9eed2216 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af7ca1e666f4434b31e3e8974aedcbf : SUCCESS in 8m 03s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/54253e2b1fa441c8a16390c04a5574cd : SUCCESS in 36m 19s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/55f1e8664caa4466a6b792da5530edd5 : SUCCESS in 39m 28s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/47ef964edebb4999b714e89da0629744 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":1},{"id":"15a5016e6ca73deda5c53618baed3f67ed2a0438","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 10:09:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ec26097c91a4cbfb9250aee3effc8c3a476f89d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 10:10:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d7cdee08414071ac4cd1116483da0b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34def1a5ee6e407f8c885e52e9f3824d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51234621087c60ef30ba6a360fe168c672af6548","revisions":{"51234621087c60ef30ba6a360fe168c672af6548":{"kind":"REWORK","_number":1,"created":"2025-05-12 10:57:11.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/86/949486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/86/949486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/86/949486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/86/949486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/86/949486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/86/949486/1"}}},"commit":{"parents":[{"commit":"2fc005e0e6af780dd7b56f026f1d0c1e2a8915c5","subject":"Remove unnecessary +x mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/2fc005e0e6af780dd7b56f026f1d0c1e2a8915c5"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-05-12 10:55:07.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-05-12 10:57:09.000000000","tz":120},"subject":"Fix lease date","message":"Fix lease date\n\nCurrently, tests are using 2026 as date in the future.\nUnfortunately, if we set the date of the machine running the\ntests in let\u0027s say 2027, then tests are failing, as per Debian\nreport here:\nhttps://bugs.debian.org/1104885\n\nThis patch fixes this by using 2046 instead of 2026 in the tests.\n\nCloses-Bug: #2110472\nChange-Id: I190daa89d134a4440dd914a9a935000aaa87d4f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/51234621087c60ef30ba6a360fe168c672af6548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/51234621087c60ef30ba6a360fe168c672af6548"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
